)]}'
{"id":"openstack%2Fmanila~985079","triplet_id":"openstack%2Fmanila~master~I9f34487173704f406424adc3c817dade0dce438d","project":"openstack/manila","branch":"master","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-04-17 10:23:13.000000000","reason":"\u003cGERRIT_ACCOUNT_38940\u003e replied on the change","reason_account":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-04-17 10:23:13.000000000","reason":"\u003cGERRIT_ACCOUNT_38940\u003e replied on the change","reason_account":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"}},"38940":{"account":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"last_update":"2026-04-17 11:15:10.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9f34487173704f406424adc3c817dade0dce438d","subject":"Fix show share instance export location ownership check","status":"NEW","created":"2026-04-17 10:21:53.000000000","updated":"2026-04-17 13:41:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c34386f373ac47f2fb3eb94edda5584e33efe97d","_number":985079,"virtual_id_number":985079,"owner":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-17 12:13:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-17 10:23:13.000000000","updated_by":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-04-17 10:23:13.000000000","updated_by":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-04-17 11:15:10.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-04-17 12:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-17 13:41:27.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"b30f8360cec73d531a26329047aa3381f0866068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"date":"2026-04-17 10:21:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93ab7c9b72be644c5887371615bdb7a44d29a0a1","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-04-17 11:15:10.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/985079","accounts_in_message":[],"_revision_number":1},{"id":"814c091d6dd12c185e28577e848e68840ab3fbd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 12:13:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ecf9176d9354b8cad8f7c9f18ce2af6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdeade17926c4866a0f85f1acc7029b5 : SUCCESS in 11m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88d02d299a4d4758aa1edc577ad31833 : SUCCESS in 35m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d21b25487b74a3d98f78f942e6bff48 : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b414de19447742b6b99475f0e4fd33f7 : SUCCESS in 8m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1792c6816be458381877d58ad61ac03 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5188c241eb714f5a8b66cbd2fbd496fd : SUCCESS in 9m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e6bbde5a96f4ca3b1f4a989274b33e6 : SUCCESS in 5m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/babe30fb6a284c6e9809d04c9200932e : SUCCESS in 1h 07m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6465140da77947a9962cd49a11b5b2b5 : FAILURE in 1h 07m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6e99f77a0c774d02827f554f3cafed39 : SUCCESS in 1h 24m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cc7effe1a49e4963a5458d347c5d2a2d : SUCCESS in 1h 03m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e74b76dcb9d14281988255d7572b3eea : SUCCESS in 1h 42m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/808372f90ef64facb8a68e1109c6d243 : FAILURE in 1h 18m 49s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0fc95c72092345b097c8ec01c263cd3c : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6214a2aee68455aaf3d695519f2d8f3 : SUCCESS in 4m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3decbf74cb4844dfa0ea0c4201d62a7a : SUCCESS in 37m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8ddd108cb0674e0e8274c52cb2483ee7 : SUCCESS in 29m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6b9c38c6deb24c74bcb0a4fd517d35d9 : SUCCESS in 38m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0f0c30d20abf408392d7de439cf5b2f1 : FAILURE in 21m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/c79efa26b8994d28853b2d935064a79c : SUCCESS in 1h 16m 54s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0bd9ec3f65454d859c3f5141661b5adf : SUCCESS in 1h 07m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3ba05b8a50404fcc8da6969cc529c1cd : FAILURE in 29m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d2c296a674124f34bfb014ba8a8dac95 : FAILURE in 28m 59s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/4de190b320434bd8bfc4627212fb00e3 : FAILURE in 4m 49s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/251b033556a3465a98acb9b342345976 : SUCCESS in 7m 33s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/92d49f368ca540de8b99f0079da3da2f : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c34386f373ac47f2fb3eb94edda5584e33efe97d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-17 13:41:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/79/985079/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/b221a32/ : FAILURE in 1h 20m 37s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/79/985079/1/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/35cb24a/ : FAILURE in 1h 25m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"478b2e5ce0c7f2f2863359da4dab909da7086101","revisions":{"478b2e5ce0c7f2f2863359da4dab909da7086101":{"kind":"REWORK","_number":1,"created":"2026-04-17 10:21:53.000000000","uploader":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"ref":"refs/changes/79/985079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/79/985079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/985079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/985079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/985079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/79/985079/1"}}},"commit":{"parents":[{"commit":"62ec140173458b4f49ba05303628b2745cb34744","subject":"Merge \"Fix flaky CephFS test_init_identify_local_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/62ec140173458b4f49ba05303628b2745cb34744"}]}],"author":{"name":"akintewe","email":"solaakintewe@gmail.com","date":"2026-04-17 10:19:36.000000000","tz":60},"committer":{"name":"akintewe","email":"solaakintewe@gmail.com","date":"2026-04-17 10:21:27.000000000","tz":60},"subject":"Fix show share instance export location ownership check","message":"Fix show share instance export location ownership check\n\nThe show() method in share_instance_export_locations did not verify\nthat the requested export location belongs to the given share instance.\nAn export location UUID from instance A could be used with instance B\u0027s\nURL and the API would return it instead of 404.\n\nPass share_instance_id to the DB query so the lookup is scoped to the\ncorrect share instance, mirroring the fix done for share export locations\nin bug #2120921.\n\nCloses-Bug: #2148658\nChange-Id: I9f34487173704f406424adc3c817dade0dce438d\nSigned-off-by: Oluwasola Akintewe \u003csolaakintewe@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/478b2e5ce0c7f2f2863359da4dab909da7086101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/478b2e5ce0c7f2f2863359da4dab909da7086101"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
