)]}'
{"id":"openstack%2Fmanila~985149","triplet_id":"openstack%2Fmanila~master~I082d40b3704d31425273abcb1df154d611ad2bd6","project":"openstack/manila","branch":"master","attention_set":{"38801":{"account":{"_account_id":38801,"name":"Simran","display_name":"Simran","email":"simran00aggarwal@gmail.com","username":"saggarwal0"},"last_update":"2026-04-17 20:28:24.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I082d40b3704d31425273abcb1df154d611ad2bd6","subject":"WIP: partial fix before deadline","status":"NEW","created":"2026-04-17 17:34:21.000000000","updated":"2026-04-17 20:28:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8ce1b24826033da4b4ac647ac3addd2ca6a85c57","_number":985149,"virtual_id_number":985149,"owner":{"_account_id":38801,"name":"Simran","display_name":"Simran","email":"simran00aggarwal@gmail.com","username":"saggarwal0"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-17 18:22:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-17 18:22:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-17 20:28:24.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"}],"messages":[{"id":"5cfad528e14970f6f38389f9d4e12180df1714d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38801,"name":"Simran","display_name":"Simran","email":"simran00aggarwal@gmail.com","username":"saggarwal0"},"date":"2026-04-17 17:34:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d009cbd5df791c557ba34357aa7cc03c39441b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 18:22:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07f327e687424acbb4d98222ce2e28ad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7869a3579bc4cd4921a2fc0ddef5718 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824926b0cd7342cabe79cd5343be7b8b : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3816b3a2dbc3415abcdacc200f2617ca : FAILURE in 5m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9abff76ac3ab434c8a2715b3e09e96bb : FAILURE in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f68435857f2e4da29ae893903b656da3 : FAILURE in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c290855947f045118bb593466c8dea69 : FAILURE in 5m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc5cb0cb8a704fb9bc333927a36185ac : FAILURE in 4m 21s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a45259fd47ff4821b5dba169d5b5b32f : FAILURE in 25m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2e5b2b8793ae413cb748cae38a8a0e44 : FAILURE in 30m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6a061c8880b2461aa3da9969a6b5a426 : FAILURE in 15m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/19ba1e4690684678b7a4449152ca7165 : FAILURE in 25m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5ad6be6e73ef4151801b8e2191917a6a : FAILURE in 31m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1b5821e239724617ba169ebc86e576e8 : FAILURE in 15m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8077b72b67c8461eb414c26a4d933627 : SUCCESS in 5m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4d186a57cac44332ba53c49688a58032 : SUCCESS in 4m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/51afa4fff2b5465aa78b405edb6531ed : FAILURE in 18m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/746b770f05ef472b8ae3536882821e72 : FAILURE in 12m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0a0519411e2a4efd964331b8cfc493fe : FAILURE in 14m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/070e263f772e40088c072bc30413460e : FAILURE in 11m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/7ae76ebbf2e3478dbca32a8cdb95d589 : FAILURE in 20m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1b9063e3be134baea087a40ca1a9b96d : FAILURE in 43m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8854e8f93cf54a2c96e543f360996d5b : FAILURE in 23m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7d6907081dfc4dc1950aa9ddad442d7a : FAILURE in 27m 10s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/ddd622968f9642b5ab6eab6f8ecfb311 : FAILURE in 6m 00s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/08a86fc10c944e988288221df3f44bde : FAILURE in 4m 57s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/8d4aba03e6864b3ca49b7714e8c33462 : POST_FAILURE in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ce1b24826033da4b4ac647ac3addd2ca6a85c57","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-04-17 20:28:24.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/985149","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf20815f43f48070675bdb4e87844c44034823b1","revisions":{"bf20815f43f48070675bdb4e87844c44034823b1":{"kind":"REWORK","_number":1,"created":"2026-04-17 17:34:21.000000000","uploader":{"_account_id":38801,"name":"Simran","display_name":"Simran","email":"simran00aggarwal@gmail.com","username":"saggarwal0"},"ref":"refs/changes/49/985149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/985149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/985149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/985149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/985149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/985149/1"}}},"commit":{"parents":[{"commit":"02a0fbfb75f099bf975485c629753f914d311daa","subject":"Add schema validation for share type show API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/02a0fbfb75f099bf975485c629753f914d311daa"}]}],"author":{"name":"Simran Aggarwal","email":"simran00aggarwal@gmail.com","date":"2026-04-17 17:32:06.000000000","tz":0},"committer":{"name":"Simran Aggarwal","email":"simran00aggarwal@gmail.com","date":"2026-04-17 17:33:33.000000000","tz":0},"subject":"WIP: partial fix before deadline","message":"WIP: partial fix before deadline\n\nChange-Id: I082d40b3704d31425273abcb1df154d611ad2bd6\nSigned-off-by: Simran Aggarwal \u003csimran00aggarwal@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bf20815f43f48070675bdb4e87844c44034823b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bf20815f43f48070675bdb4e87844c44034823b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
