)]}'
{"id":"openstack%2Fmanila~985151","triplet_id":"openstack%2Fmanila~master~I0646bad255d9dba2d61c34b6fef07bf217c6bf4a","project":"openstack/manila","branch":"master","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-04-17 18:02:22.000000000","reason":"\u003cGERRIT_ACCOUNT_38940\u003e replied on the change","reason_account":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-04-17 18:02:22.000000000","reason":"\u003cGERRIT_ACCOUNT_38940\u003e replied on the change","reason_account":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"}},"38940":{"account":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"last_update":"2026-04-17 20:39:18.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0646bad255d9dba2d61c34b6fef07bf217c6bf4a","subject":"Fix show snapshot instance export location ownership check","status":"NEW","created":"2026-04-17 17:59:53.000000000","updated":"2026-04-17 21:17:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2acd7d571722582d1545596b38742b0b9b394d8e","_number":985151,"virtual_id_number":985151,"owner":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-17 19:09:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-17 18:02:22.000000000","updated_by":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-04-17 18:02:22.000000000","updated_by":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-04-17 19:09:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-17 20:39:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-04-17 21:17:10.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"}],"messages":[{"id":"6a3e6cbc0970414877559e673dc81d32a5863b8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"date":"2026-04-17 17:59:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bda0bbac4c713028f8b109f51e81f67abe96999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 19:09:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f06507514f04f4c99151da930f0f636\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de321c923f5a4b94a730282ee7bff289 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e734d3a4f7d42df86fe50562aa99881 : SUCCESS in 18m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a449a1a8a4743f4b14d1f85745f3b57 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/195db2d7af7647c5b87affcbe4e9a62f : SUCCESS in 8m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36176d45da2540da8928f64489f30c56 : SUCCESS in 5m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96727965e06e40d6a7bbfa2ac4913348 : SUCCESS in 14m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af4392924093420c86f101be2faa3a53 : SUCCESS in 8m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3fce1574562145afb4fc08a9333f86cd : SUCCESS in 59m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5c60dfed0b9a4cf48aff15377597b30a : SUCCESS in 51m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f3a07940422c4fc086edfaa380076390 : SUCCESS in 1h 06m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/59993a09c1094ad78a6fb797bf894b43 : SUCCESS in 41m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/898b33a6eba0461586781f295763e23d : SUCCESS in 52m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5b37f1e94dd64ea5bc6a01fa676cabcd : FAILURE in 39m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/64aa919519d142048c3f4da62d2013e6 : SUCCESS in 2m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f4cb730145b9464490254037c5ebfaaf : SUCCESS in 6m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/224b3c97723446078ec2a480fe097ad7 : SUCCESS in 30m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cf67097eb68b4099bea89f84008c50f3 : SUCCESS in 23m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3434c6f188414a27b7e506f79ba8334e : SUCCESS in 30m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7392781d0c0c4758a742f3f575a5492a : FAILURE in 17m 51s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/c2ac1d119e4c4b71befac3c382c2effd : SUCCESS in 1h 01m 34s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7f287f9e74f84a9ca7d9bf56999b9343 : SUCCESS in 52m 40s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b9d8fde8304147168bed54d8709cbe27 : FAILURE in 13m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9a72d231569d48c3a2792ad6feec874f : FAILURE in 13m 32s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/0db921ebea9c47149a233c9afe0ad367 : FAILURE in 2m 47s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/85a377f3d282446b930b8d9de6bf9348 : SUCCESS in 5m 07s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/384a7fb1d0c64512a8ae0ca51a9729f1 : SUCCESS in 7m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52765fbf14d86251f3a9b4f5b2c8f6fec08f5114","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-17 20:39:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/985151/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/08c3029/ : RETRY_LIMIT in 7m 43s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/985151/1/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/60f018c/ : FAILURE in 1h 27m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2acd7d571722582d1545596b38742b0b9b394d8e","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-04-17 21:17:10.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/985151","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"567b00b02efa24a03c461d69d10803659285feb4","revisions":{"567b00b02efa24a03c461d69d10803659285feb4":{"kind":"REWORK","_number":1,"created":"2026-04-17 17:59:53.000000000","uploader":{"_account_id":38940,"name":"Oluwasola Akintewe","display_name":"nathan_akin","email":"solaakintewe@gmail.com","username":"nathan_akin"},"ref":"refs/changes/51/985151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/51/985151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/985151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/985151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/985151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/51/985151/1"}}},"commit":{"parents":[{"commit":"62ec140173458b4f49ba05303628b2745cb34744","subject":"Merge \"Fix flaky CephFS test_init_identify_local_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/62ec140173458b4f49ba05303628b2745cb34744"}]}],"author":{"name":"akintewe","email":"solaakintewe@gmail.com","date":"2026-04-17 17:59:07.000000000","tz":60},"committer":{"name":"akintewe","email":"solaakintewe@gmail.com","date":"2026-04-17 17:59:07.000000000","tz":60},"subject":"Fix show snapshot instance export location ownership check","message":"Fix show snapshot instance export location ownership check\n\nThe show() method in share_snapshot_instance_export_locations did not\nverify that the requested export location belongs to the given snapshot\ninstance. An export location UUID from snapshot instance A could be used\nwith instance B\u0027s URL and the API would return it instead of 404.\n\nPass snapshot_instance_id to the DB query so the lookup is scoped to\nthe correct snapshot instance, same fix as bug #2120921 and #2148658.\n\nCloses-Bug: #2148704\nChange-Id: I0646bad255d9dba2d61c34b6fef07bf217c6bf4a\nSigned-off-by: Oluwasola Akintewe \u003csolaakintewe@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/567b00b02efa24a03c461d69d10803659285feb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/567b00b02efa24a03c461d69d10803659285feb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
