)]}'
{"id":"openstack%2Fmanila~985233","triplet_id":"openstack%2Fmanila~master~Icfdec06980b851fba7a75de123c220a01ceb5cc0","project":"openstack/manila","branch":"master","topic":"bug/1998489","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-05-13 16:47:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icfdec06980b851fba7a75de123c220a01ceb5cc0","subject":"[DNM] Patch os-ken spawn_after for BGP stability","status":"ABANDONED","created":"2026-04-18 17:44:16.000000000","updated":"2026-05-13 16:47:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"916af70073b505748f3042dbe0760b43f4b7f41f","_number":985233,"virtual_id_number":985233,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-18 19:13:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-18 19:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-18 19:24:10.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"}],"messages":[{"id":"a802869386ef55a8e33a420c2039c71b5e7cef61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-18 17:44:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a5811450de9fd8a2d8d5391104939de08f0df8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 19:13:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0640502ca3c8408eac35d6a1eb37f527\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87caf73a97ec429fbd637ac20d06982e : POST_FAILURE in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c81ae87a1b464656baf9835d07955957 : POST_FAILURE in 22m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0947825040641d0a320d2e367997f33 : POST_FAILURE in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c0c9aff028d4f3686f081be4a7c180b : POST_FAILURE in 8m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a04076d8350e41fcab884999783529b2 : POST_FAILURE in 7m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/651eb1e6889b4a5aa5b56fdc18da9f90 : POST_FAILURE in 8m 51s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ab28ef7e159c4189bbdccbf5f228a92a : POST_FAILURE in 7m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8350aef4ca3b4b36897fd3079bc50002 : POST_FAILURE in 55m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fdb0d0665f844d80a7cc6b17ee37232c : POST_FAILURE in 39m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/146ea3e168e6488b9232c6a70e2f39d5 : POST_FAILURE in 52m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8c9ba7c6082e4defbb63a3c215ae6678 : POST_FAILURE in 1h 03m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/51b9d3529ad649d98f255da61d3e0f31 : POST_FAILURE in 51m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d4b4039c44ac42b49b5aab5e5aa9a2a1 : POST_FAILURE in 3m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b1d487000b97425dbc048950c540dde7 : POST_FAILURE in 4m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/135b61a9999946e08214806711066340 : POST_FAILURE in 34m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f9ea08efb8e142c8ae5ba21a998db593 : POST_FAILURE in 24m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3e7d7ef6078a4878ba60b5ec606c4cd2 : POST_FAILURE in 38m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/74c5bbfc977b45c4a76d42e3a74442bf : POST_FAILURE in 14m 45s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/ee3f3c1d704e44779752d1a2e1a56702 : POST_FAILURE in 1h 23m 22s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3b448800442141f39ea29bcb5861a213 : POST_FAILURE in 17m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/de25447d030a4206af8767abaccc35d0 : POST_FAILURE in 14m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b4ee958ec96240b98f5f4e4d4d1ba19a : POST_FAILURE in 19m 04s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/120723e4143e4db78826019c95a3f66b : POST_FAILURE in 4m 05s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/8d7863dc5bdb4fc185f730a234017ee9 : POST_FAILURE in 9m 04s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e50db1542c1040579205974cddba0219 : POST_FAILURE in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd063eb4c216473ee9f66515d267f7feb3276625","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-04-18 19:24:10.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/985233","accounts_in_message":[],"_revision_number":1},{"id":"916af70073b505748f3042dbe0760b43f4b7f41f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-13 16:47:10.000000000","message":"Abandoned\n\nos-ken\u0027s fix is now available in 4.2.0: https://launchpad.net/bugs/2148744","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9e859d3f413f9ca7ba90870fca1007058067837","revisions":{"a9e859d3f413f9ca7ba90870fca1007058067837":{"kind":"REWORK","_number":1,"created":"2026-04-18 17:44:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/985233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/33/985233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/985233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/985233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/985233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/33/985233/1"}}},"commit":{"parents":[{"commit":"9880395371db6ad9ad031d9019716d21cb1a34a7","subject":"Replace Quagga with FRR in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9880395371db6ad9ad031d9019716d21cb1a34a7"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-18 17:40:54.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-18 17:40:54.000000000","tz":-420},"subject":"[DNM] Patch os-ken spawn_after for BGP stability","message":"[DNM] Patch os-ken spawn_after for BGP stability\n\nWork around LP#2148744 by patching os-ken\u0027s native\nhub spawn_after to use threading.Timer, whose\ncancel() actually works. Without this, BGP hold\ntimers fire regardless of keepalive resets, causing\nsessions to flap every 30 seconds.\n\nChange-Id: Icfdec06980b851fba7a75de123c220a01ceb5cc0\nRelated-Bug: #2148744\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9e859d3f413f9ca7ba90870fca1007058067837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9e859d3f413f9ca7ba90870fca1007058067837"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
