)]}'
{"id":"openstack%2Fmanila~990144","triplet_id":"openstack%2Fmanila~master~I2a5af39e8be5044f0a47f6b6c2c1c2db2503d8ee","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-01 17:55:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a5af39e8be5044f0a47f6b6c2c1c2db2503d8ee","subject":"devstack: Refactor library installation from source","status":"MERGED","created":"2026-05-26 17:02:15.000000000","updated":"2026-06-01 17:56:58.000000000","submitted":"2026-06-01 17:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990144","meta_rev_id":"2c352f6557616007616a349cac3bf9f27049d0ee","_number":990144,"virtual_id_number":990144,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-01 17:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-06-01 16:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-06-01 16:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 18:11:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-27 13:10:27.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-06-01 16:39:56.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"58eb4b44f462e9b1290507d20f6aad5adaac632b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-26 17:02:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec184024ce013f9f564b8d1de916e00c8fc5e61e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 18:11:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56fe9585bdc545f19a82bbffadee5cf0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec4bae9e86445d3a387430d853f1ae7 : SUCCESS in 10m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/541cf8b7bfe74b3fbd20953284186e7f : FAILURE in 17m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5efd43926b413ca46190cf1da7e7a1 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e996e7460394ee88eda21b903096826 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de53d420c5764a559a2f53f8732c2101 : SUCCESS in 9m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe90bf3fd3424cc28624653ff65017b2 : SUCCESS in 6m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03c503f3610541e7b6332a2169280a78 : SUCCESS in 9m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4c1c1bcb00074e04b20f8c0a79c7f243 : FAILURE in 10m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/176bb6838ee94cd9bb08d4b2ad8528e7 : FAILURE in 8m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e9b0f9dc8cf942c3b9832c48a61afbb8 : FAILURE in 14m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5d01752cb19e48f39249e3a37d963f2d : FAILURE in 18m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d707dcf61592444fb3f67456fbef264b : FAILURE in 12m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a5028d200ad042c4aa35092a8b4b7f28 : FAILURE in 8m 04s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/945c86b6224649e8b4bd08a1a295be99 : SUCCESS in 4m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9bbd1037d2f34782bd36afd36aefc9e5 : SUCCESS in 3m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/27b3ffd751e04aa8955bbdc2d4df3701 : FAILURE in 16m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9a5b2af1217d499ca5130e18ada2c2e7 : FAILURE in 16m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/66738ba9634544b7b9eeacb2ebd1d680 : FAILURE in 6m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1d3ad540904f43e5a7dfd7461575c5c6 : FAILURE in 8m 03s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/2ee22932105e4c74a2b92d829cb7e57b : FAILURE in 36m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d98eb49ea8b948008ff5a0c98d67e514 : SUCCESS in 1h 01m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/006b8463eaba4e2d8178069d6ca59a84 : FAILURE in 8m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/822a795cffbb482aa09efb87d94b78f2 : FAILURE in 18m 59s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/4811b6293c39402898a389c6043ef1dd : FAILURE in 4m 11s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/e6cb8145406444c889d729cc242dbe31 : SUCCESS in 8m 07s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/62609cffa65a4a09a09cef5c548031ad : SUCCESS in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0db00748a03d79a2368afcb69d9a82fe227cbc99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-27 03:00:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"91a1587122710c6d4cdb9e7b5b3491850feebbcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 04:08:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/747d914b0bf34a6abe9e952fe90fa2c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01bdcd89bbdb47c3985ead8993c05a37 : SUCCESS in 15m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28550a1efbf14c5a9d213fa453f555bf : SUCCESS in 37m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeb20d1b6a414af5a1c4fc6c13ef0877 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d502cf4f10774329aec257e8253703da : SUCCESS in 8m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7244a5b8d3674620992ea0adb2c89bcb : SUCCESS in 8m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/72540965724149cd9bbe5fe7cbad46b4 : SUCCESS in 6m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5625ea2d8254c7f9000db63b0232ebe : SUCCESS in 9m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/433edd406fcb40d29e485efbc5b9d94d : FAILURE in 15m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7df5b35681904b8b9dab8b6506148eee : FAILURE in 16m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4cb418fb8f2e4e92a06b4d86a7fd4a54 : FAILURE in 21m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/80583a4e999e42559e7e3cf2f037691a : FAILURE in 20m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0180d91024614e9889c8d5d577d8fedc : FAILURE in 25m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f58c2ff5f9934078a9efea035d995381 : FAILURE in 18m 58s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e76971d581ee471c908e06981de45f01 : SUCCESS in 4m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/70bc52527d524f43a4f37a4098253742 : SUCCESS in 4m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/388b87160d0d45bd8db217248230ae2e : FAILURE in 9m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c62c2ad5c0bd4e2ca201bfc85f57c7bd : FAILURE in 7m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/945be06e0ba5409ebf01795a39500ebf : FAILURE in 11m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/14a91705f1a74efcbab5c104603d3ce0 : FAILURE in 10m 54s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/dc9cdb3f158648738f76069cd63242d4 : FAILURE in 32m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/02ff6b68ea6142bf9c6e254d8d7ff7a9 : SUCCESS in 1h 03m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1a59c9e9f6274b31b51fc5f542d7c9ac : FAILURE in 19m 43s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/58bc75d282cc4e50ba19a34f92896f27 : FAILURE in 20m 11s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/d9e1f8e1a7874726bc53e3ad85caa7b7 : FAILURE in 4m 46s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/3666f8b2ba274c08a9c3b3b7bd0dbb37 : SUCCESS in 8m 32s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1a71dcb619c149d39b60e1efd21d0c10 : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6d4094c17bb63225f40009b70c1e7e4051c1502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-27 09:45:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d1b681a267e2e6ca979e4ef5c673ddb83b373c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 11:54:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b775d0b0994141d7b03bf793155a53e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a145e8b501cd4084ac0550dbf9dc2b40 : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76560005aebf41ca8b92820d750492bb : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/432c0edb022841ce8db6b472fb1f299d : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05dd94b7455b4bc68c20074c42122782 : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcbd35d13b4046f3adc8714ccc31690c : SUCCESS in 6m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dee8ca653dcf400a9cb7998ebb486568 : SUCCESS in 11m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/390f921318634264893b873b9a234e3d : SUCCESS in 4m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/441888350a6946fbb1385391b2f3e38d : SUCCESS in 1h 13m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8a713e46101e468a93b73d43d44690fd : FAILURE in 2h 03m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/99bf3ac77c26423f82afab810c144f90 : SUCCESS in 1h 30m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/663a33c80a724c73b834ea65bc2e0b95 : SUCCESS in 46m 31s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2e1d6b23a7a341b685f199be6bc09a13 : SUCCESS in 1h 14m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ed9772a57a1f4d8e8f05184a41fc30ff : FAILURE in 52m 42s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cc7d2dcd3d584a9089cad29bd5daec8e : SUCCESS in 2m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b52b2a01d161490aa251f253e0d3ae1c : SUCCESS in 2m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b7558843e45f43afbc5769a4a23d45df : SUCCESS in 21m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/fd21f63264ba4e1ca04b55f9982d7381 : SUCCESS in 32m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ef01f3b92f3745f8bb7a74c99822a5bb : SUCCESS in 34m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f458bc4f9c914590a2a61fde1c90e92a : FAILURE in 22m 26s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/bd7b9ef25bd24bddb425eadb0cb9ffd0 : SUCCESS in 59m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e433c38c6c0c4b88bc939541d01be05f : SUCCESS in 45m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8e79b9b97f3e43d2a67517be0952c695 : SUCCESS in 24m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/905f773b589d484d8d95154fab3fa470 : SUCCESS in 20m 29s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/cc2947036ebd49b4a504fd30497e1d77 : FAILURE in 4m 43s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/db784ccdcc61442ea873588b8505d5e6 : SUCCESS in 8m 44s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7c01ac1ba54740ce9039e9295a1de3e4 : SUCCESS in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d184e6202c6882b6bbecbd37f27f3deb22183f09","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-27 13:10:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/990144/3/upstream-check-manila/manila-tempest-plugin-ontap-dhss/18defaa/ : SUCCESS in 1h 13m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"9050b0709dc8598af6d457397be986fe3bd1fa68","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-01 16:39:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"899f35a981d50273e57e8ea92de494adb298ff7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:40:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4d616fcde6b6195363bc27076c30c2082b47ebb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 17:55:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/303b1729d2bc4706b6b02c86e869fa3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00656ae91d5d43709b2e3f271e150d97 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92aad212b97b4077b156ddef565c964e : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/739e2e5cccf24c1c8200bfaeb2f762f9 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce54f8ed1e6a483a97018b45bca3e368 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/416f89f4ebf24b2eb5cbb86b9981f4c0 : SUCCESS in 9m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fb43dbb575e74a978a71024b40a8279d : SUCCESS in 1h 09m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/031f8cb6f5fe4c82a12a35e0d8a70afb : SUCCESS in 35m 16s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bbc6ec399ab6497a87a0a5daafcf07cb : SUCCESS in 33m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5315f8c64c104c54ad72e857ddd687d4 : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":3},{"id":"3422ee9e9b27297dc89c28dcd0232404ff1c3de3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 17:55:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2c352f6557616007616a349cac3bf9f27049d0ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 17:56:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/666aad19383448cc84fafbeac3d671b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3a9ff5e1cc4942953f5ebac3a0496e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c9fb66b9aea2dfae1cea544d6c3c8e1b3013fcdf","revisions":{"51c591b11cd41e36dd9648d7d7d1ed3277d74a92":{"kind":"REWORK","_number":1,"created":"2026-05-26 17:02:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/990144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/990144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/990144/1"}}},"commit":{"parents":[{"commit":"7187dafcc160f894a6a7152a6561994fbcfc6d0c","subject":"Use systemctl command to manage service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7187dafcc160f894a6a7152a6561994fbcfc6d0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-26 17:01:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-26 17:01:15.000000000","tz":540},"subject":"devstack: Refactor library installation from source","message":"devstack: Refactor library installation from source\n\n... using the common utilities from devstack.\n\nChange-Id: I2a5af39e8be5044f0a47f6b6c2c1c2db2503d8ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/51c591b11cd41e36dd9648d7d7d1ed3277d74a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/51c591b11cd41e36dd9648d7d7d1ed3277d74a92"}]},"branch":"refs/heads/master"},"8b151dfee11ab781a61c4953e7b575172a9c0c85":{"kind":"REWORK","_number":2,"created":"2026-05-27 03:00:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/990144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/990144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/990144/2"}}},"commit":{"parents":[{"commit":"7187dafcc160f894a6a7152a6561994fbcfc6d0c","subject":"Use systemctl command to manage service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7187dafcc160f894a6a7152a6561994fbcfc6d0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-26 17:01:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-27 03:00:46.000000000","tz":540},"subject":"devstack: Refactor library installation from source","message":"devstack: Refactor library installation from source\n\n... using the common utilities from devstack.\n\nChange-Id: I2a5af39e8be5044f0a47f6b6c2c1c2db2503d8ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8b151dfee11ab781a61c4953e7b575172a9c0c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8b151dfee11ab781a61c4953e7b575172a9c0c85"}]},"branch":"refs/heads/master"},"c9fb66b9aea2dfae1cea544d6c3c8e1b3013fcdf":{"kind":"REWORK","_number":3,"created":"2026-05-27 09:45:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/990144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/990144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/990144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/990144/3"}}},"commit":{"parents":[{"commit":"7187dafcc160f894a6a7152a6561994fbcfc6d0c","subject":"Use systemctl command to manage service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7187dafcc160f894a6a7152a6561994fbcfc6d0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-26 17:01:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-27 09:45:20.000000000","tz":540},"subject":"devstack: Refactor library installation from source","message":"devstack: Refactor library installation from source\n\n... using the common utilities from devstack.\n\nChange-Id: I2a5af39e8be5044f0a47f6b6c2c1c2db2503d8ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c9fb66b9aea2dfae1cea544d6c3c8e1b3013fcdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c9fb66b9aea2dfae1cea544d6c3c8e1b3013fcdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
