)]}'
{"id":"openstack%2Fmanila~993473","triplet_id":"openstack%2Fmanila~stable%2F2025.2~I7428a9f5af0bc887ac484fcfa76c12987c99e7f9","project":"openstack/manila","branch":"stable/2025.2","topic":"backport/tune2fs","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-06-16 15:25:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7428a9f5af0bc887ac484fcfa76c12987c99e7f9","subject":"Fix volume reserved space issue by adding tune2fs","status":"MERGED","created":"2026-06-15 23:37:50.000000000","updated":"2026-06-16 15:25:52.000000000","submitted":"2026-06-16 15:25:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993473-backport/tune2fs","meta_rev_id":"d979e8c1f40e48ff4ddc0cc79734290c61d8bfb3","_number":993473,"virtual_id_number":993473,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 15:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-06-16 12:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-06-16 12:07:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-15 23:37:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"state":"CC"},{"updated":"2026-06-16 01:24:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 02:44:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-06-16 12:07:36.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"6561d35229b2072306a337086a4c89b683171949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-15 23:37:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa753441789c0e65baee732298ccfa5ab238a4b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 01:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02b09e8f08974bbb97067ab145874cf3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ae894a63ea4e6fb4d642cb0d07d8b0 : SUCCESS in 13m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae58bb2d5c5a4c1899e2b8a4b1ea13c0 : SUCCESS in 35m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b8db02c9cd481caea71a84d131538b : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11e0bf95446b4599ad79a077c10ebc2d : SUCCESS in 10m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a738fc17ec7e4c8d9634a66c08c11d5e : SUCCESS in 8m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e078af334dea4023a9705cae4e7df7bb : SUCCESS in 15m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7be7e803e67944a1ab26b3c6bc15f03d : SUCCESS in 5m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9c2eb61e37844ff09977c2adcf3c2c5b : SUCCESS in 1h 12m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/af5e0edfba534a48aea6f48664725327 : SUCCESS in 1h 36m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/756ecc00232c4abdae832c42560ff1d9 : SUCCESS in 1h 09m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f5d3221a37d5464e9605bee0c35cb488 : SUCCESS in 56m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5a91569972bd474d90d5576279b5ccf2 : SUCCESS in 1h 40m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cc0bbcf1f6ce47848b5fa7858e4432b4 : FAILURE in 16m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7790ec1679b14fcbaa9e014721662640 : SUCCESS in 4m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e090f38d0294d69b48572cd4f765c18 : SUCCESS in 4m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/46d0e0d7d8884294ad87cac3b7bf9a88 : SUCCESS in 34m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/41cc93bf5b4c49a1bf5a1a44757b4661 : SUCCESS in 38m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d0147cf4ff2e44f990135e06d034eda1 : FAILURE in 15m 46s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/2f306ac7400b4c2680d939d092c05be6 : FAILURE in 18m 25s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a819df42b05f4cf09c0b065a3b8039a0 : SUCCESS in 1h 12m 30s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a38f192090b4ded837ce032f2e1337b : SUCCESS in 53m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/84c24243c4394985a38d91d906054ef1 : SUCCESS in 34m 05s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/022d763f548e4c8b884697c460cb0158 : SUCCESS in 33m 18s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/cec57c39aa56460597dcc9b8197f8bb7 : FAILURE in 7m 19s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1d7bebb1c0bd4c6bbcb75c66730edd9b : SUCCESS in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3a711028f1ae766b74528be8c3d392b462657a1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-16 02:44:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/993473/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/beaa1e6/ : SUCCESS in 1h 17m 38s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fdf2e496d7ea8941d23cd7733725a878225db446","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-06-16 12:07:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab8363c2862b0502efc59e6cb9c064d82262fd14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 12:08:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"536bfdb781f7ffa3633b4f2bde13e372992fb2cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:25:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/455b3e373e2c48c2a001baea9f6c8a99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3223097af1db4f1c9b5f6d7bea73e264 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2ff6dd58395425db575504acffe6372 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/372d700c93de45d1b5dcf5f984c24608 : SUCCESS in 9m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b966e9975f742b5a40cd6893470aa1c : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33f519ad1e924079a7946fbd2a0e8a1c : SUCCESS in 9m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5b76bffa2d624b6db46ed49f17853e10 : SUCCESS in 1h 16m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a21b56210090425b99610dac801d99ed : SUCCESS in 33m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ecd9429889bf461e8dc41904db7dcc1f : SUCCESS in 47m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d979e8c1f40e48ff4ddc0cc79734290c61d8bfb3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:25:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eaf3a6624751b875b46f691ab8e20ba59d804a7a","revisions":{"eaf3a6624751b875b46f691ab8e20ba59d804a7a":{"kind":"REWORK","_number":1,"created":"2026-06-15 23:37:50.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/73/993473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/73/993473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/73/993473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/73/993473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/73/993473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/73/993473/1"}}},"commit":{"parents":[{"commit":"4aa5b735c61973c11e1d594a5ec3ad0c2d8f63a8","subject":"Derive Barbican endpoint path from service catalog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4aa5b735c61973c11e1d594a5ec3ad0c2d8f63a8"}]}],"author":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-10-30 04:19:57.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-06-15 23:37:44.000000000","tz":-420},"subject":"Fix volume reserved space issue by adding tune2fs","message":"Fix volume reserved space issue by adding tune2fs\n\nWhen using the generic driver with Ceph-backed Cinder volumes,\nthe NFS mount size displayed on the client is slightly smaller\nthan expected. This occurs because ext-based filesystems reserve\na small portion (typically 5%) of disk space for the root user.\n\nFix: a new configuration option is introduced:\n`set_zero_reserved_blocks`. When enabled (default True), the\ndriver executes a `tune2fs -m 0` command after formatting ext\nvolumes, ensuring the full capacity is available to the share.\n\nThe `_format_device()` method was updated to include this step,\nand appropriate warning handling was added in case the tune2fs\ncommand fails. The existing `test_format_device` was updated to\nreflect the new command sequence, and new tests were added to\nverify different filesystem and configuration scenarios.\n\nTest Plan:\n- test_format_device: Verifies both mkfs and tune2fs commands\n  are executed for ext filesystems when the flag is enabled.\n- test_format_device_runs_tune2fs_for_ext_filesystems:\n  Ensures tune2fs runs for ext filesystems and mkfs is executed.\n- test_format_device_skips_tune2fs_for_non_ext:\n  Confirms tune2fs is skipped when using non-ext filesystems.\n- test_format_device_skips_tune2fs_if_config_disabled:\n  Checks that tune2fs is not executed if the config flag is false.\n- test_format_device_logs_warning_if_tune2fs_fails:\n  Verifies that tune2fs failures are logged as warnings without\n  causing the operation to fail\n\nCloses-Bug: #2060776\n\nChange-Id: I7428a9f5af0bc887ac484fcfa76c12987c99e7f9\nSigned-off-by: Rose Kimondo \u003crosewairimu1@gmail.com\u003e\n(cherry picked from commit 2d3773e44d4221d931c0aeb4aae41f1da193798a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/eaf3a6624751b875b46f691ab8e20ba59d804a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/eaf3a6624751b875b46f691ab8e20ba59d804a7a"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
