)]}'
{"id":"openstack%2Fmanila~993476","triplet_id":"openstack%2Fmanila~stable%2F2025.2~I8a7fa3a764c993b4ec4f0bcfff40bc5b63c2c298","project":"openstack/manila","branch":"stable/2025.2","topic":"backport/pagination-links","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-06-16 15:26:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a7fa3a764c993b4ec4f0bcfff40bc5b63c2c298","subject":"Fix pagination links with incorrect endpoint names","status":"MERGED","created":"2026-06-15 23:38:48.000000000","updated":"2026-06-16 15:31:16.000000000","submitted":"2026-06-16 15:26:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993476-backport/pagination-links","meta_rev_id":"fe22d22a4e9be66e03ab538e49e3689cb5f5756d","_number":993476,"virtual_id_number":993476,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 15:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-06-16 12:08:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-06-16 12:08:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-15 23:38:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"state":"CC"},{"updated":"2026-06-16 01:46:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 03:01:53.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-06-16 12:08:53.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"b833ed61b4be6fa3ba3c570a26db230d0dc1030c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-15 23:38:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8569d7799803046837f94a850cb6944062df8f66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 01:46:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbf0178653414713ad084a60f181d3f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9fb5e8a221407ab65aaeeae3623846 : SUCCESS in 12m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7226fc20a7964b9d98b4c6f4d6aa6071 : SUCCESS in 31m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a85046cc804342aead9438f551f19724 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e63ce8442dd5479da555bfa81fcb3b94 : SUCCESS in 9m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1032a594ee5f44139b16f6e169e739d4 : SUCCESS in 10m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aee681deac0f47c3b2a873356778e924 : SUCCESS in 8m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb813e048df046229f1803c27ecb85b9 : SUCCESS in 9m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f4b9f08c85be4093a24368f6fdc9b1a8 : SUCCESS in 1h 17m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3e2dceaa447b469c980591538801c577 : SUCCESS in 2h 01m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/63c7b549ab774ab393c2ae663a3ae245 : SUCCESS in 1h 33m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5cfd8f63ee60434cacf79876a77c3a6e : SUCCESS in 42m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/928c4dc9d91b46fc8e017a9b0e9f5f50 : SUCCESS in 1h 20m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7e3b79a355094076a1be861aeea7a7ea : FAILURE in 11m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d1b63d96b024439b914900582430e7eb : SUCCESS in 3m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/109cade9a9544d21a0faf5d380743cec : SUCCESS in 4m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f1a9f293f2f5467c879587cd8255386e : SUCCESS in 26m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1486bb1f0d144a838dcbff8dd7beb185 : SUCCESS in 32m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ac10723d6b9143c09bfe0d40551d0568 : FAILURE in 11m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/fa5f74d8252645acb352fbd0e101e777 : FAILURE in 14m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/77674f5fb9aa48e4aff42aa67128a7f6 : SUCCESS in 44m 42s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98b5edfd0f5a487ab03b547c9c984051 : SUCCESS in 46m 37s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f666fe23329d41f99aa16e3833a43e0e : SUCCESS in 20m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1737b27cfcbc4d049a50b458701a1a28 : SUCCESS in 21m 34s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/e2d62bf49bad4bf98dd24e52b7e7d6a9 : FAILURE in 3m 20s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f287ed9db04c42119337344e99e0c5dd : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd377f487e162d459d4a2abd3d2412e0e48e8f6b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-16 03:01:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/993476/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/144a6a8/ : FAILURE in 1h 12m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f0e2106a8122349a13c1eb9ce570816d303f0312","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-06-16 12:08:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e38c32d3b61caa357f10059294eb9397167a563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 12:09:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04b41e1a6c627e52b223b9a9701a0fabc1e63119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:26:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e602e4a4652a4cb285d0e67fa62a5074\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9079652b787d4a15913cddec08ca4869 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2444023772724a93baf5707ce8f62b21 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/119252f5038540c7a31d6c4d8ea5bce3 : SUCCESS in 12m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e3a46589aff4b3dba8db0e4432203cf : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5dfc2a4ac9240dbadc1ce6d42b44506 : SUCCESS in 9m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/37c9d164294540b5bdb96f838c8718b9 : SUCCESS in 1h 16m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a4e62ff986d749afafd2ac6321e2efdb : SUCCESS in 22m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ae143d6ab97e4409b53a806213865658 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":1},{"id":"32f34bf53b08bf352ac18c04d3b6b85324aff2cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:26:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fe22d22a4e9be66e03ab538e49e3689cb5f5756d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:31:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb9869a980ff45298635ec580ffcfdd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b976c8999840b8b83e85157f87d9c4 : SUCCESS in 2m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b564520c7a314df1b7927955fd228fae : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a86a6186dc5e4202bc43be4c4e56825c50d1cec","revisions":{"0a86a6186dc5e4202bc43be4c4e56825c50d1cec":{"kind":"REWORK","_number":1,"created":"2026-06-15 23:38:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/993476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/76/993476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/993476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/993476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/993476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/76/993476/1"}}},"commit":{"parents":[{"commit":"4aa5b735c61973c11e1d594a5ec3ad0c2d8f63a8","subject":"Derive Barbican endpoint path from service catalog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4aa5b735c61973c11e1d594a5ec3ad0c2d8f63a8"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-22 14:58:51.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-06-15 23:38:45.000000000","tz":-420},"subject":"Fix pagination links with incorrect endpoint names","message":"Fix pagination links with incorrect endpoint names\n\nManila\u0027s API has a problem generating pagination links\n(\"href\" field of \"self\", \"next\", \"bookmark\" items) when\nthe collection name (for example: \"share_groups\") is different\nfrom the route name (for example: \"share-groups\").\nThis patch fixes the broken endpoints by introducing a _collection_route_name\n\nPartial-bug: #2058427\nChange-Id: I8a7fa3a764c993b4ec4f0bcfff40bc5b63c2c298\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n(cherry picked from commit 6ab2a1e59bc0f8980f53ff4c04736310a5290ec8)\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0a86a6186dc5e4202bc43be4c4e56825c50d1cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0a86a6186dc5e4202bc43be4c4e56825c50d1cec"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
