)]}'
{"id":"openstack%2Fmanila~993477","triplet_id":"openstack%2Fmanila~stable%2F2025.2~Id49a8e8f0e014e2f6ea0bfe41750ebe9a6f96509","project":"openstack/manila","branch":"stable/2025.2","topic":"bug/1824442","attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-06-16 03:30:50.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Id49a8e8f0e014e2f6ea0bfe41750ebe9a6f96509","subject":"Remove privileged fields from networks APIs","status":"NEW","created":"2026-06-15 23:39:00.000000000","updated":"2026-06-16 03:30:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e57411044688a2103cc0977b7e9bebf6a84a8b17","_number":993477,"virtual_id_number":993477,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-16 02:00:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":-1,"date":"2026-06-15 23:40:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-16 02:00:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 03:30:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"3a68f17ca3553e5cb6592ee3e4fd583d4ed49191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-15 23:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7858ee3e1065c8bc31b70f4afb012bee1ce3e9c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-15 23:39:30.000000000","message":"Topic backport/network-privileged-fields removed","accounts_in_message":[],"_revision_number":1},{"id":"84a5eb0c550599b4739414789c0ff669b9bbe6a1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-15 23:39:38.000000000","message":"Topic set to bug/1824442","accounts_in_message":[],"_revision_number":1},{"id":"14f7013944d63f50b9c3a93bbc3c2caddcd88024","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-15 23:40:01.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"997261c055e467a21015cb3bf0b88892fe02c3f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 02:00:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0101fcdf754476bc208d4db04890c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a20f6fe3d84367b7c12339f8dcdaf8 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19a7857293c848e1b78ae72f9d4a3664 : FAILURE in 35m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ed9cc2b3c14414a794883762118487 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bc8ebd28c4747d0abf89f780a7c218b : SUCCESS in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b71a69ea561e4858b81be37fac406a0e : SUCCESS in 7m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bce4e21bfb9c4806bcecf4eb80760a98 : SUCCESS in 14m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02be69de62d04c198d271a3c063e1aa8 : SUCCESS in 9m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/71cbc00c6e8b4e998522538a33d4c45f : SUCCESS in 1h 10m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fd599b98c236431390fd033f715de7ed : SUCCESS in 2h 09m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/84ea812c625f4821abdeb8c83eb7876e : SUCCESS in 1h 10m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6594fcf782f747178dcf4b51739c6fff : SUCCESS in 42m 16s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/73f7cd41339347f6a838b8042318814f : SUCCESS in 1h 32m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8da68f6c02d44ccba8fb67b53b725643 : FAILURE in 22m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8bc9f17860d347fc9461daee470b9c1c : SUCCESS in 5m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/70e9d7c76bdf488a98a01d9fb3fa66ce : SUCCESS in 4m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/11fa6df6e70744c0bf2cb81961deeaba : SUCCESS in 33m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/50640ac12a7344c490a968e60651a440 : SUCCESS in 29m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/07053339dcd94c10bd086f5970e92f4c : FAILURE in 10m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/074a930397764f8ca43fa1c3c7e92478 : FAILURE in 9m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/db3164d6b49c49cc85c905dfd1f91903 : SUCCESS in 45m 20s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/630bb331e2a54f8ca547fdb8e26ed0c9 : SUCCESS in 1h 01m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4266c8d48a444d6f850155bebe3e7f84 : SUCCESS in 20m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8270ec1471e24fba9bae0a9b26c8ff72 : SUCCESS in 31m 32s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/ba6c485d8bd84c55941eb0747c88a582 : FAILURE in 4m 27s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/be771a389b0a40c3840c39bb88ec411d : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e57411044688a2103cc0977b7e9bebf6a84a8b17","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-16 03:30:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/993477/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/bb757fa/ : SUCCESS in 1h 11m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b24f0dee9a037b14ea25b60023c1c046d3ecaed","revisions":{"3b24f0dee9a037b14ea25b60023c1c046d3ecaed":{"kind":"REWORK","_number":1,"created":"2026-06-15 23:39:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/993477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/77/993477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/77/993477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/77/993477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/77/993477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/77/993477/1"}}},"commit":{"parents":[{"commit":"4aa5b735c61973c11e1d594a5ec3ad0c2d8f63a8","subject":"Derive Barbican endpoint path from service catalog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4aa5b735c61973c11e1d594a5ec3ad0c2d8f63a8"}]}],"author":{"name":"FelipeLooze","email":"luizfelipe.looze@gmail.com","date":"2022-08-01 13:02:52.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-06-15 23:38:54.000000000","tz":-420},"subject":"Remove privileged fields from networks APIs","message":"Remove privileged fields from networks APIs\n\nAdded a new policy to guard fields \u0027segmentation_id\u0027\nand \u0027network_type\u0027 in the responses of share network\nAPIs. These fields are  nullified when APIs are\ninvoked by users that are not admins by the\nvirtue of default RBAC. These users can\u0027t use these\nattributes to filter networks either.\n\nCloses-Bug: #1824442\nAssisted-By: Claude Opus 4.6\nCo-authored-by: Lucas de Oliveira \u003clucasmoliveira059@gmail.com\u003e\nCo-authored-by: Riane Torres \u003criane.torres@gmail.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/979050\nChange-Id: Id49a8e8f0e014e2f6ea0bfe41750ebe9a6f96509\nSigned-off-by: Riane Torres \u003criane.torres@gmail.com\u003e\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit bf3c835c94c997048a46389fba1bfd9c4f55da27)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3b24f0dee9a037b14ea25b60023c1c046d3ecaed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3b24f0dee9a037b14ea25b60023c1c046d3ecaed"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
