)]}'
{"id":"openstack%2Fmanila~993570","triplet_id":"openstack%2Fmanila~master~Ie743d7293179f975d7aa0628d2bc741e25170ebe","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2026-06-19 11:04:07.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-06-19 11:04:07.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-06-19 11:04:07.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-06-19 11:04:07.000000000","reason":"Change was submitted"},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-06-17 06:10:44.000000000","reason":"\u003cGERRIT_ACCOUNT_38371\u003e replied on the change","reason_account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-06-19 10:27:58.000000000","reason":"\u003cGERRIT_ACCOUNT_36180\u003e replied on the change","reason_account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-06-19 10:02:59.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"39161":{"account":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"last_update":"2026-06-19 11:04:07.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-06-19 11:04:07.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-06-18 16:31:45.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}}},"hashtags":[],"change_id":"Ie743d7293179f975d7aa0628d2bc741e25170ebe","subject":"Netapp: Removed wrong field ip_addresses from get_cluster_peers in REST","status":"MERGED","created":"2026-06-16 10:53:39.000000000","updated":"2026-09-21 05:00:28.000000000","submitted":"2026-06-19 11:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993570","meta_rev_id":"f159f64eef47b043756dd5237b13c9d912c1c4a5","_number":993570,"virtual_id_number":993570,"owner":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-19 11:04:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"recommended":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"all":[{"value":1,"date":"2026-06-17 06:10:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":2,"date":"2026-06-19 10:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":2,"date":"2026-06-19 10:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2026-06-18 16:31:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"all":[{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":1,"date":"2026-06-19 10:02:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-16 10:54:30.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-06-16 10:54:30.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-06-16 10:54:30.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-16 10:54:30.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-16 10:54:30.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-16 11:20:04.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-16 13:01:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 15:28:25.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-06-17 05:11:54.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-06-17 05:11:54.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-06-17 05:11:54.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2026-06-17 05:11:54.000000000","updated_by":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"14eb05a9b0f8a72a1104c1ced9c670899c1060c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"date":"2026-06-16 10:53:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d2761aa8aba51598b9521d106b6ec7b1b394e0c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-16 11:20:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2d4fcd88def94af0b54a62acec7c7e6b\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/85993a0a2fb648649fde4a04574e7cf2 : SUCCESS in 25m 53s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/13f5a024505449aab03cd1e23ec33e13 : FAILURE in 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29a56a4f31fb45ded8a465c361645ef5632513d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 13:01:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4edde82dc0ed4389bef37585a8b1106e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c746bf99ab34807bac9188abbdc233e : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c42985e639f64e0799d3add2a12e47ef : SUCCESS in 35m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d29fd7646c848679d273298c92dd2e5 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7060f22bfdce4e188fdcd96a29cc8fa2 : SUCCESS in 10m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b07fd44e07344087b6b474cb957f8f84 : SUCCESS in 9m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3aca4bb0e2a94cdd8c250bc3f6546880 : SUCCESS in 11m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e9bc0cfbc9d4d8fa97a3119c79b7907 : SUCCESS in 4m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/74ab1d96e53f477d93b165e40446982f : SUCCESS in 1h 07m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ecea70f6106d47fcb8f377bafee391c9 : FAILURE in 2h 05m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bc3708dfaaff459ab7188fae948d346c : SUCCESS in 1h 05m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/38656980dfea487c90d75d235e2c8399 : SUCCESS in 32m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e8d9cd9bbca540b19762873466562378 : SUCCESS in 1h 16m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d4e5436327df4002a01c3f31d1ca4e72 : FAILURE in 39m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a57d6a0ee4614c0bb6fbc63550cbf368 : SUCCESS in 5m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd1e29663c5b48b88082034c240f459e : SUCCESS in 4m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4e6e2c46437f4627a860e52004bce448 : SUCCESS in 29m 13s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ad19e1cf608f410b834416c5745a7f7a : SUCCESS in 31m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/02ff758c00aa43e2a4c33d1b335b189f : SUCCESS in 39m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9de91b7050a545039c38fa7f059041b9 : FAILURE in 11m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/6db577a285404d1791c28f2cee4d1c27 : SUCCESS in 1h 30m 26s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/fe9692c21ade4ed6a17c4a1e65946aa4 : SUCCESS in 43m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/44098b17501f4dacb596c8a5f151bc68 : SUCCESS in 32m 49s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f58a8be8a3b24117979868a799b778b4 : SUCCESS in 35m 47s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/0219553f3dbc48e3b28a187bd0f1472e : FAILURE in 2m 45s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/26194bc502054137a12392613ecd51f1 : SUCCESS in 8m 59s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3ad2a31ac1ee4c1d949e2f5a3cbf9a99 : SUCCESS in 3m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8776e1896a9f1481d7dcd9765746ed39db6f6f9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-16 15:28:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/993570/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/271450f/ : SUCCESS in 1h 14m 21s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/993570/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/577422b/ : FAILURE in 1h 23m 44s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f039e7efe8fde2d3c3c1695d2b5a257a1c69fbcc","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-16 16:11:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf3f2afb7d557e979cdf835ad92218418a76a3a7","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-06-17 05:14:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03de22abb54cff931511cca3f00be39290285e8d","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-06-17 05:20:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8d44030678e51ce9a6d5e5a7be5e0813267c604e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"date":"2026-06-17 05:42:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4f42fa67a6204678607b373ad89ae2e0add7b52b","author":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"date":"2026-06-17 05:43:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52c429574cfe63a7fa723a5db57840c78a9068ee","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-06-17 06:10:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36aad28afcdf79750663fa7e2486a28e3aff1f25","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-17 06:15:39.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/dfc00b933636477ba22cef7522402b1a\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aafc6da54cdc40f6962bacf41cf07445 : FAILURE in 32m 01s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6f1681a3b81247faad5025728dae88a2 : FAILURE in 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e223303c1242f377946ebd4beb42cebe90828bf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 07:10:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3a4c5aeacc342cba396ead1a46de298\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbde46c3107b4d198c857eb3cbd37ac0 : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/167320cd94cf4e02bd920cee9353f1ba : FAILURE in 31m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb91310341114c02b16f2b1ca5ef5a39 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfc4103c172d43449341adc84f12b0af : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bdd730f856b14bc7ba0512935c51501b : SUCCESS in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4d98a2a65d3245cdb52b716c66272b03 : SUCCESS in 11m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2adda474be33424ebf3dc589be5b76e7 : SUCCESS in 4m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e505c367761c4e2598557a1156d31790 : FAILURE in 27m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ab98a640f8244670b8c9a671773832d7 : SUCCESS in 56m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b7d7c7bba8334dcf85e95a29c151bda2 : SUCCESS in 1h 00m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/39032c3d681f4e1cb45ca8ba5bff1be4 : SUCCESS in 39m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/84d8b4ef8bd24a9c85f1f8d9d1c6dd5f : SUCCESS in 1h 22m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6f8e8f4a6da4436aa946f05156ad0cdc : FAILURE in 58m 10s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e5b17d10c74458b934a07e272785567 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77dae871ee324893819203739a0d8640 : SUCCESS in 4m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e1e87e9280f8415b84fde43063e99088 : SUCCESS in 23m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6c4b79bd29c44871ad80c7297ae1edca : SUCCESS in 22m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/139e8f8a92e34cb7a7295261850daf8d : SUCCESS in 40m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0972285e7d23460ebc4c2a939addac55 : FAILURE in 18m 18s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/df150a7a2c364e56a4e2fb80e529ab64 : SUCCESS in 49m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b0aac400ff674aff90b08649ecd4f892 : SUCCESS in 1h 05m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2e6b859fc96b4982bb944c5573c5cf51 : SUCCESS in 36m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e9adc69daceb42f3965733c18be4d3bc : SUCCESS in 24m 17s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/615ba655b19a4bdaa91a94b5c8492b4c : FAILURE in 4m 27s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/594f87d2c0b84499a40ead2daee56336 : SUCCESS in 8m 42s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/87bd421f22ed4a23b7e09b2b1ddaa892 : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b7e3778922b6e8b6fddd5cadc63892ce4c91349","author":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"date":"2026-06-18 04:14:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0febcc1fc91fe7adb8e8f01e2363d424f1967531","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-18 04:45:58.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/91cc828a99b44e0daaf4e55113f9ac63\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c64197eca2dd473d8a9f5fbed8f82811 : FAILURE in 30m 27s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ce40c2c4d814cdcb24b0cacd3b4e8c9 : FAILURE in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cc627ecaa32c83b58d088cbb1ec2a34c86d60b6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 05:51:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcbea62b9f934834b7a17b1580405f29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5587fa1b944a4e13908127b876df3d2a : SUCCESS in 5m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba7b4cf2bd0c4637a557dcc9abec30eb : FAILURE in 17m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3188661703b84831b19dd999e69c2416 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10203dfbd9a64680970af71386323511 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15382fd86cdb47d68b60d579544fd354 : SUCCESS in 6m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46b5ffafd3b5459f8fab1918abcf70ab : SUCCESS in 12m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9fd27be9da14fca98973f7e9c1334fc : SUCCESS in 5m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/691ab3a857b04bf39e878c393dc14dcb : SUCCESS in 1h 21m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9efe69bb18f1442d84a63663cb26cd4a : SUCCESS in 56m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/95525ab979db43ccb0711993002f15e1 : SUCCESS in 1h 23m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7a8cb441fbee4313a1bec8375c5949f2 : SUCCESS in 47m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f8e22dc9b4784f0390f6564c4b3d2987 : SUCCESS in 1h 26m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4d131a5c38254d3a9a228942976bfa51 : FAILURE in 1h 26m 38s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/132e4531840f45bb9594ab5ed881fb39 : SUCCESS in 4m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/03c5dc7cb0ad47c1952f8c90d46c3197 : SUCCESS in 3m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ba771c11fc554f58bcf83f2fdd5a87a9 : SUCCESS in 29m 53s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/30d8adcf01e5417ba02f1b730612aaa6 : SUCCESS in 31m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/48583fcbf9f94f8285784bb6e2dc6208 : SUCCESS in 46m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cb6e5ace5a8c4827ac6c239a96e36f2a : FAILURE in 16m 25s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/fecb28bf7f7240149bf7a17746f33095 : SUCCESS in 31m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/12b6318fa20c4474b4a832a2ff4d070f : SUCCESS in 53m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/293c4468ce8243cba50613ed16d0ed43 : SUCCESS in 34m 10s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/28e7fc1db21b4c0482f598062c93b240 : SUCCESS in 35m 00s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/7f9eae0a4e64469e9536fa8e8984f0cd : FAILURE in 4m 24s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/c8def55eb114497d81458d046e90c45a : SUCCESS in 9m 07s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/8c391cf1e4bf41689df0eda0aa793fa3 : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"909ca5019ec837c558f55b8cd725184d61aea902","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-18 07:27:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/993570/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/17c147e/ : SUCCESS in 1h 16m 58s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/993570/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/9f6f31b/ : FAILURE in 1h 34m 38s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"aa4c229e0db64221b85678d4e18e0ea73d1d768b","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-18 16:31:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c70c0556e94ef3055fb800d416c4fcf2cb377d4d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-06-19 10:02:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b6fac254ec4085dbb2cddf7a6815a3f0e74e761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 10:03:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8de504c5dd19c428c5ccbec2ca10ab5124a2fd06","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-06-19 10:27:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"222b03f2289ac3dd287ddb3a6df62fe9db538bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 11:04:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f30c403f693474b94d26805b9f5f8fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bf93e6d5cf4c42b6d37ce4c7cb675e : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9df033fda4694bba8dac7c48b5947994 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6269a5c2a1944462a3093d8525da4885 : SUCCESS in 7m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac8f0ba95abd4f1aa98fe50ad2832b9e : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab47c347b547462dbabf95eb2caaebc2 : SUCCESS in 9m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3835c391708a4b92ba5a4e8a25ad861d : SUCCESS in 58m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/78283c62946b492db7f9739c904287ef : SUCCESS in 23m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/10ff6cc0885b449f9e7bde0971971a08 : SUCCESS in 19m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/875a96651afa4cf798fc4ef200fbacff : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":2},{"id":"47a6af7ad5add7fbab5d4e09e63c96c2a4f49974","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 11:04:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d84c22601b1a5cd164801c629b3cab3d73285f65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 11:05:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/769b18a993644969984cd357c634177a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe64d57a39548e18db97f24e8e73817 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2},{"id":"55fadef3628e744efb4eaad5c04d58ded9974171","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-19 03:19:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f159f64eef47b043756dd5237b13c9d912c1c4a5","author":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"date":"2026-09-21 05:00:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30b4f66de77d7950a96882d71f2958ede120a477","revisions":{"d033507013b8b497e446395f2b81c5f406a55855":{"kind":"REWORK","_number":1,"created":"2026-06-16 10:53:39.000000000","uploader":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"ref":"refs/changes/70/993570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/70/993570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/993570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/993570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/993570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/70/993570/1"}}},"commit":{"parents":[{"commit":"5cf57fbe77c3ede8914294524fefd4f2134c2d40","subject":"Merge \"Netapp Rest implementation for get_cluster_peers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5cf57fbe77c3ede8914294524fefd4f2134c2d40"}]}],"author":{"name":"Medha Choudhary","email":"medhac@netapp.com","date":"2026-06-16 10:51:35.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhac@netapp.com","date":"2026-06-16 10:53:14.000000000","tz":-240},"subject":"Netapp: Removed wrong field ip_addresses from get_cluster_peers in REST","message":"Netapp: Removed wrong field ip_addresses from get_cluster_peers in REST\n\nCloses-Bug: #2156875\nChange-Id: Ie743d7293179f975d7aa0628d2bc741e25170ebe\nSigned-off-by: Medha Choudhary \u003cmedhac@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d033507013b8b497e446395f2b81c5f406a55855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d033507013b8b497e446395f2b81c5f406a55855"}]},"branch":"refs/heads/master"},"30b4f66de77d7950a96882d71f2958ede120a477":{"kind":"REWORK","_number":2,"created":"2026-06-17 05:42:59.000000000","uploader":{"_account_id":39161,"name":"Medha Choudhary netapp","display_name":"Medha Choudhary (netapp)","email":"medhac@netapp.com","username":"medhac","status":"Netapp"},"ref":"refs/changes/70/993570/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/70/993570/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/993570/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/993570/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/993570/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/70/993570/2"}}},"commit":{"parents":[{"commit":"5cf57fbe77c3ede8914294524fefd4f2134c2d40","subject":"Merge \"Netapp Rest implementation for get_cluster_peers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5cf57fbe77c3ede8914294524fefd4f2134c2d40"}]}],"author":{"name":"Medha Choudhary","email":"medhac@netapp.com","date":"2026-06-16 10:51:35.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhac@netapp.com","date":"2026-06-17 05:40:20.000000000","tz":-240},"subject":"Netapp: Removed wrong field ip_addresses from get_cluster_peers in REST","message":"Netapp: Removed wrong field ip_addresses from get_cluster_peers in REST\n\nCloses-Bug: #2156875\nChange-Id: Ie743d7293179f975d7aa0628d2bc741e25170ebe\nSigned-off-by: Medha Choudhary \u003cmedhac@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/30b4f66de77d7950a96882d71f2958ede120a477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/30b4f66de77d7950a96882d71f2958ede120a477"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
