)]}'
{"id":"openstack%2Fmanila~995000","triplet_id":"openstack%2Fmanila~stable%2F2025.2~Iadefd06f68abab4b6996c2b527ed64934e1cb34a","project":"openstack/manila","branch":"stable/2025.2","topic":"bug/2142898","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-08-24 19:29:40.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-08-24 20:35:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadefd06f68abab4b6996c2b527ed64934e1cb34a","subject":"Adopt to new castellan","status":"MERGED","created":"2026-06-26 05:18:17.000000000","updated":"2026-08-24 20:37:20.000000000","submitted":"2026-08-24 20:35:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995000-bug/2142898","cherry_pick_of_change":978319,"cherry_pick_of_patch_set":13,"meta_rev_id":"ac67db26ebb59cacf69329d3ca72578ec70ea6be","_number":995000,"virtual_id_number":995000,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 20:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-08-24 19:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-08-24 19:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 05:18:17.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-06-26 05:47:25.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-26 07:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 08:54:08.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-08-23 07:36:09.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"25f85cadc0727e85116cd90c6a09fd021609ac88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-26 05:18:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0815a1e90a4a9c8f2ebf762801f5c567b853a82","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-26 05:18:43.000000000","message":"Topic bug/2142898-stable/2025.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"1edf06dbfd5b591ade5c3744443cb004cd8c0348","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-26 05:18:47.000000000","message":"Topic set to bug/2142898","accounts_in_message":[],"_revision_number":1},{"id":"b3eaddeb9edde5a2d05ac4125d6979193a43a49f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-26 05:47:25.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fd4b9d0ec587480599d2bf6babac626e\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/985aecacee5b4d80998c0b3493cdd48e : FAILURE in 28m 18s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfb35c575b3a4244975d8ee9049b6e4b : FAILURE in 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebde329042d8e35e87931241c86380aaaa7e9f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 07:28:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0ee9faff80045329e802e95e261df85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3176d8e072b407dbf75482286d4f590 : SUCCESS in 9m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fab87e3584b428fb22ab1497910e200 : FAILURE in 21m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4aa0c9298c44296871840e72dc86823 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65ec1a3c92524aa9a454abf77ce7b09f : SUCCESS in 7m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d6ac519488b443088f04c73e0691155 : SUCCESS in 6m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/62821d4718ad4392ba2331f970ecff1d : SUCCESS in 10m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d3930e20d0a41848ea9f6c82d8cfb71 : SUCCESS in 9m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f3d3a218f99e44f79a42404f2b4e9c43 : SUCCESS in 1h 07m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7502f2bf8d7343d3afddad89b8b6b15d : FAILURE in 2h 09m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/01b422606d0348b8b3c9a51bb661f0ab : SUCCESS in 1h 07m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4d3105b7a8c7470481c5654f828c4b3e : SUCCESS in 28m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dbc21332ee464a1d8efa064792b61ac8 : SUCCESS in 1h 35m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f47959d332fa4c068a82500fbd4065da : FAILURE in 9m 28s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe3641eb44804ac286834cfabcd1deb0 : SUCCESS in 6m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bcb457216b84417fafb0900da1aa6c21 : SUCCESS in 4m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/31ca2263ba2d419a92c3f7d3a7e7d313 : SUCCESS in 30m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b0577303e0e8434ba9e082fb9157d437 : SUCCESS in 33m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/60a228080449419da6df8bb7ea4e87a7 : FAILURE in 8m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/2e65c58658cd45c388d1d6a24374843f : FAILURE in 18m 38s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d74d9c8619fc41dcb5b7f29048c631a9 : SUCCESS in 53m 31s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/633916ee5ea84ed88518ad20803a3733 : SUCCESS in 1h 03m 58s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9463955421f74e3b8f0a31fb6b1c62b3 : SUCCESS in 13m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7cf7b139c0fd4b6dbf8cbc2d93670ffe : SUCCESS in 32m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/bd4e151941bb43329f7f56196cd0b5a5 : FAILURE in 5m 44s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/78f6b7e484954a43855919cf71233fa7 : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87658a7a12b2c014142bb86b893aad9208afc778","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-26 08:54:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/995000/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/4bd69c8/ : FAILURE in 1h 24m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"638534b08fbc6c55ffeeb86acee0e7e8dc4ef025","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-08-24 19:29:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40f4b11b4853287f1d5a63c30ad8dffda01220a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 19:30:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8664bc11c00c79b372967fcbc5b617738fa77410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:35:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a83a5f969f15495ba8e7d7867785beb6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1135e2841b4a4ce1be6696a4c7477974 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17af046e437649269fa6846199e7b25a : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a5b9af998cd4b4e817158b459525bc0 : SUCCESS in 11m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80141a599fec44759419211b506e765c : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dabd25ab8d1439ba9783fda7b41b455 : SUCCESS in 9m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5b2f9b6e84cd4aedb1e4a4158e1cf3d2 : SUCCESS in 1h 00m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a2c912d4cd304c3d963f5f37d5c904f4 : SUCCESS in 20m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/92f32bbc67c740ba9e2ff1feab9a29b2 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b15015d5bef17b7f45d90eb049730095442a0da8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:35:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac67db26ebb59cacf69329d3ca72578ec70ea6be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:37:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6283a362d7741c68fdd65535062fb62\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98613eda25f24bd185cbba20c2e2ae88 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccdae85e9dc404cf8285b90dbf00ddc02c9c38e0","revisions":{"ccdae85e9dc404cf8285b90dbf00ddc02c9c38e0":{"kind":"REWORK","_number":1,"created":"2026-06-26 05:18:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/00/995000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/00/995000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/995000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/995000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/995000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/00/995000/1"}}},"commit":{"parents":[{"commit":"e514c58c2787aea479495baca7de34d818cbaf0d","subject":"Merge \"Fix pagination links with incorrect endpoint names\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e514c58c2787aea479495baca7de34d818cbaf0d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-06-26 05:18:17.000000000","tz":0},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is more robust than implementing endpoint detection in\nmultiple layers, and allows us to ensure that the requested secret can\nbe accessed early.\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nCloses-Bug: #2142898\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit c10154bc7997dbdef7b50ebb3f5a1b13b28118c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ccdae85e9dc404cf8285b90dbf00ddc02c9c38e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ccdae85e9dc404cf8285b90dbf00ddc02c9c38e0"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"9f499ff74881daef98502ced89377791eb990a6e","ours":"e514c58c2787aea479495baca7de34d818cbaf0d","theirs":"c10154bc7997dbdef7b50ebb3f5a1b13b28118c7","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
