)]}'
{"id":"openstack%2Fmanila~997487","triplet_id":"openstack%2Fmanila~master~I092b58b2e3051d1f2d98451c4bfd1768ef087933","project":"openstack/manila","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-10-02 13:29:14.000000000","reason":"A robot voted negatively on a label"},"38702":{"account":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"last_update":"2026-07-20 17:23:31.000000000","reason":"Some votes were not copied to the current patch set"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I092b58b2e3051d1f2d98451c4bfd1768ef087933","subject":"Maintain driver dependencies","status":"NEW","created":"2026-07-16 08:05:17.000000000","updated":"2026-10-02 13:29:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d7212fe85559fe339165b3d404bddf8920875a57","_number":997487,"virtual_id_number":997487,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-10-02 13:29:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-16 09:08:54.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-16 10:07:14.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-07-16 10:26:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-16 14:45:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-07-20 17:22:52.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-10-02 10:42:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"ecc8f4dd0d8d7bb2ea965e826fc7ff6036c53195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-16 08:05:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"302627efebcdf74f06785acd8d7ecc6f327b93e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-16 08:08:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9dad0562b3325f60bb9c7b91c6ed72fac2c606e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-16 08:14:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7dbdbb34f8d2ad458727e8b8c5d2a898505b3b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-16 08:14:52.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e0b831963d095d1cdf0f6e04da812cab499db72c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 09:08:54.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/bff6608da5914e4d9f696add870d7ca5/","accounts_in_message":[],"_revision_number":4},{"id":"129246d61e46dba6278ddc3f801c41641278b0de","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 09:18:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/bff6608da5914e4d9f696add870d7ca5\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dd41012eb3f84d1095fa4e49c662d0c4 : SUCCESS in 31m 40s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2c3c8ac3635d4133b1f1d09a817ba2b0 : FAILURE in 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bedaa97974b028d50ab20cc5c72c871cf9b0062d","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-07-16 10:07:14.000000000","message":"Patch Set 4:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/997487","accounts_in_message":[],"_revision_number":4},{"id":"8ca464503e72d732066161505f7157099c78df51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-16 10:10:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"84a3e130361dc7fcfefdd418053643783a9d44c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-16 10:20:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a07ee67d6c4311ab014cb5f46c51151ccbb4e042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 10:26:55.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/requirements/+/997489 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"061b659e7b8c5eed7ad18da2b05c984da444acaa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 10:26:59.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/requirements/+/997489 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"e89249398dd946c809a8527683b10cd408267259","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-16 10:27:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52b94db7381386f49ca4e589c5fb0fa5044433ed","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 11:40:22.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/aed98b291e2441a9871142c2d50215ec/","accounts_in_message":[],"_revision_number":6},{"id":"1e40c260ffdd11f349154d719bc9be2da32c0a9a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 11:47:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/aed98b291e2441a9871142c2d50215ec\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d7d02bade49c4ddab0abf91daa05ff1d : SUCCESS in 30m 31s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a90a1934aefc405bb047a10ad647d9c8 : FAILURE in 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5fdc3670125e7a49666cc9937551bf0cae2c80b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-16 12:04:16.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"7bc1f1548c54c6906c3ecfc89863864c0cee952f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 13:28:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba40297e388c49699b4692f600c4ab97\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63f1e088bd994467bc14a0fa09da6979 : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9aea57a3994ece9ce0852f27f89b69 : SUCCESS in 12m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e723856a9b7488c804c56fd9ff5747e : SUCCESS in 30m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c5f756088242459c3cb331c9a8778a : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74480cc2401d40348c56c3c6add41b8e : SUCCESS in 9m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68c94b36884549d389b47a857b41f490 : SUCCESS in 9m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4fa18cf67c141a193a56a03353d5063 : SUCCESS in 11m 39s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/788e230f50304e45af83c48b5d2114da : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ff5c1efb0214d2987cd2dfbde3c3b46 : SUCCESS in 8m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bd5c5df4c4b14cd9bcedf1ba10bf05bb : SUCCESS in 1h 14m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/793d88054a654a59afa74ccf63692791 : SUCCESS in 47m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/de543b5d5646421f86473c225ab2ad43 : SUCCESS in 1h 14m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/093b46318a8e4b14b74c2d7afa35986b : FAILURE in 31m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9b739db761f74fafa87af9a17a6c2321 : SUCCESS in 1h 08m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9744510b0ec54a2db77cc27172d4b276 : FAILURE in 36m 04s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3add332db6bf4d038ae658714e336eb7 : SUCCESS in 4m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f24890339b149e3aeefce3b3268d7b0 : SUCCESS in 4m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e212b82527ad46bc9e0dd70553f37e59 : SUCCESS in 38m 03s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9eaf3dfbecd34894b88c491cb094c5f2 : SUCCESS in 23m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2905c6fbb0714945aa4b393d8e4f040d : SUCCESS in 41m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f028ad207ce947c89e4282c8fccb35f0 : FAILURE in 48m 12s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/6ac56531a81b4af7b1540ab1cae4f816 : SUCCESS in 1h 13m 24s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/41f1bfd311bf4816a2fc3f7a81b82b71 : SUCCESS in 1h 15m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e0c2c69537974db490f962bdd20e4af8 : SUCCESS in 36m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2293291c7fa74de098c9da6416c007ba : SUCCESS in 40m 14s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/0ed68596c3ea4e8a9d5cbbe4b4284241 : FAILURE in 2m 35s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/97378f9803fa472a9689d25f8dc9e882 : SUCCESS in 8m 47s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1af0fd41449e47bbb6a8d377c7aebfdf : SUCCESS in 7m 57s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"2074fa7b334e9edce7ee2483af29ae04c1276605","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 13:56:07.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/3adc8f13f2d7465ab618e9b179a931a7/","accounts_in_message":[],"_revision_number":7},{"id":"5e03bc3f224ffd361242e657ee28224a43cb6f79","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 14:11:59.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3adc8f13f2d7465ab618e9b179a931a7\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/73b4c78352f1407980de1fb033538ce9 : SUCCESS in 32m 12s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/44dcea15ba3c40fc8d4205b1f69721e9 : FAILURE in 19m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"933c60d80d28c0746e93604aeb39460c20f3a2cb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-16 14:45:17.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/997487/7/upstream-check-manila/manila-tempest-plugin-ontap-dhss/02834b7/ : SUCCESS in 1h 14m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"583fab17a65c29654f615a33579705e083cc0b9d","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-07-16 16:29:56.000000000","message":"Patch Set 7:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/997487","accounts_in_message":[],"_revision_number":7},{"id":"788e94abb0b8e68a2b147b38e4ceb7bba7cc8c4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:22:52.000000000","message":"Patch Set 2:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/145/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"add94cfe5acc24e317a4ff3cb69e1bdde460c807","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:22:54.000000000","message":"Patch Set 4:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/147/ : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"0fb03dc70dac89333dbdc74d52d3f796add3efb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:22:56.000000000","message":"Patch Set 5:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/149/ : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"b99ee849c2cd826dfa8b1e6622c1ae4f00b02b07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:23:02.000000000","message":"Patch Set 7:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/154/ : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"27cd98c05d75604ed194a3c844904f92a82c94d4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:23:31.000000000","message":"Patch Set 3:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/146/ : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"d8da47f855556336f0bb9c0f858061c067f2320f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:23:35.000000000","message":"Patch Set 6:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/151/ : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"110db52276e08f3a6f6a32dcc77dce7d8ded072f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-10-02 10:42:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b301fdfe67a9fa5528bee875c6b2c03c6fdda778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-02 10:50:41.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"17455a3931314631de5de533f90e35d682bf482b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-02 10:54:57.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"cd597ad799effebc83653eb082cc5406c3a363a0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 11:34:50.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/64115faa8e7c42b19d77a257f6f71d3e/","accounts_in_message":[],"_revision_number":9},{"id":"26d59106b8f3167d1b9cb34847c735b296c6b64d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 11:59:34.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/64115faa8e7c42b19d77a257f6f71d3e\n\n- pure-devstack-manila-tempest-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb474121c27540de81ca3163de44e1ed : SUCCESS in 33m 02s\n- pure-devstack-manila-tempest-fa-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f8e6ecc3ca064d4e81f389fa9852d558 : FAILURE in 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d7212fe85559fe339165b3d404bddf8920875a57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 13:29:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc5fd98ba90f437d8aec2c6b11e41aad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/689523900efd4c66ae7e7ef7bbbf2a58 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2570821bc9451eb3683d36e4f41828 : SUCCESS in 11m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d589dc030d64ef49c323b00f8f02b2a : SUCCESS in 33m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/101b82a584a043e8a0ce7c346ea942a4 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c67f4b9f51b24af3849db7f1b23bc2d1 : SUCCESS in 6m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a23580b518147d787fa295e426fae88 : SUCCESS in 13m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aff053e8287f4c1080fe3eb926119478 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11ea785ce1d5428aa8e6b4135c3d610a : SUCCESS in 9m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4afba0ce43e24dae911d46d2bd39513f : FAILURE in 25m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cb1ef8b3ea9347f9997cb29a710b3c31 : FAILURE in 2h 20m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/56f06bb024264960bb7bf70456a58bd1 : SUCCESS in 1h 22m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/aa63153c30db4566bf5013d33312d10a : FAILURE in 40m 28s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c091715abc78458880e81e1bf0d376fb : FAILURE in 30m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1418190b1e094e73904a3d22afbc25e1 : SUCCESS in 1h 14m 45s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/e377c01bd5c1465fbe03c2a73333bb0e : SUCCESS in 1h 07m 44s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7fa4428f3366476fb2ae12b147c03914 : SUCCESS in 4m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5066c21add8e41d8902c8134a7b61851 : SUCCESS in 5m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/618c8feedbe84f0799f41cc55f60b850 : SUCCESS in 37m 06s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/a8ae8d72a70c43d59135feead8c373b2 : SUCCESS in 33m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/871c6345929e462a89c72b313f28d5fe : SUCCESS in 25m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b014eccb16cc409f9efa208b73c60040 : SUCCESS in 59m 12s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/4d536f845f11490b8d88f6aa1ce11b76 : SUCCESS in 1h 17m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/01f3f43a683b40e886d51f613864fe00 : SUCCESS in 59m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b92646021deb4dce9ebdbb57451dad26 : SUCCESS in 33m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ea06cb6f5ead41438d5583d397f8e112 : SUCCESS in 18m 28s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/adcf3a6c91e64e03b526db92933794de : FAILURE in 4m 49s (non-voting)\n- manila-tox-py3-eventlet https://zuul.opendev.org/t/openstack/build/85456dc97a924f6583ca2481e264db8f : SUCCESS in 5m 32s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/53babc9ef9fc44a0b2cd8b49736cf249 : SUCCESS in 7m 21s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c0bf83c7e10d615d10dfd31cb4e4772626050fce","revisions":{"8a6b699dd6d941d9aa209fedb2b243ddb4f4b577":{"kind":"REWORK","_number":1,"created":"2026-07-16 08:05:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/1"}}},"commit":{"parents":[{"commit":"06d83802ac613c93dd95435ac3ab7d778aaebd39","subject":"cephfs: Remove deprecated cephfs_volume_path_prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06d83802ac613c93dd95435ac3ab7d778aaebd39"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nAlso simplify how the extra dependencies are handled for Infinidat\ndriver, following the existing implementation in their cinder volume\ndriver.\n\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8a6b699dd6d941d9aa209fedb2b243ddb4f4b577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8a6b699dd6d941d9aa209fedb2b243ddb4f4b577"}]},"branch":"refs/heads/master"},"ae31ae9029e7ae91023b84142d2748cea2fe32cb":{"kind":"REWORK","_number":2,"created":"2026-07-16 08:08:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/2"}}},"commit":{"parents":[{"commit":"06d83802ac613c93dd95435ac3ab7d778aaebd39","subject":"cephfs: Remove deprecated cephfs_volume_path_prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06d83802ac613c93dd95435ac3ab7d778aaebd39"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:08:26.000000000","tz":540},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nAlso simplify how the extra dependencies are handled for Infinidat\ndriver, following the existing implementation in their cinder volume\ndriver.\n\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ae31ae9029e7ae91023b84142d2748cea2fe32cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ae31ae9029e7ae91023b84142d2748cea2fe32cb"}]},"branch":"refs/heads/master"},"87efaeed79a66f7d67e3e86b2345d0210876515e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-07-16 08:14:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/3"}}},"commit":{"parents":[{"commit":"06d83802ac613c93dd95435ac3ab7d778aaebd39","subject":"cephfs: Remove deprecated cephfs_volume_path_prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06d83802ac613c93dd95435ac3ab7d778aaebd39"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:14:13.000000000","tz":0},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nAlso simplify how the extra dependencies are handled for Infinidat\ndriver, following the existing implementation in their cinder volume\ndriver.\n\nNeeded-By: https://review.opendev.org/c/openstack/manila/+/997487\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/87efaeed79a66f7d67e3e86b2345d0210876515e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/87efaeed79a66f7d67e3e86b2345d0210876515e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"57de4b5e174d0341109ea5237f351fc4ead21bed":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-16 08:14:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/4"}}},"commit":{"parents":[{"commit":"06d83802ac613c93dd95435ac3ab7d778aaebd39","subject":"cephfs: Remove deprecated cephfs_volume_path_prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06d83802ac613c93dd95435ac3ab7d778aaebd39"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:14:52.000000000","tz":0},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nAlso simplify how the extra dependencies are handled for Infinidat\ndriver, following the existing implementation in their cinder volume\ndriver.\n\nDepends-on: https://review.opendev.org/997489\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/57de4b5e174d0341109ea5237f351fc4ead21bed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/57de4b5e174d0341109ea5237f351fc4ead21bed"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5b5739b69719c98bf2a742ca8955ce7a87677c8a":{"kind":"REWORK","_number":5,"created":"2026-07-16 10:10:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/5"}}},"commit":{"parents":[{"commit":"06d83802ac613c93dd95435ac3ab7d778aaebd39","subject":"cephfs: Remove deprecated cephfs_volume_path_prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06d83802ac613c93dd95435ac3ab7d778aaebd39"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 10:09:38.000000000","tz":540},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nAlso simplify how the extra dependencies are handled for Infinidat\ndriver, following the existing implementation in their cinder volume\ndriver.\n\nDepends-on: https://review.opendev.org/997489\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5b5739b69719c98bf2a742ca8955ce7a87677c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5b5739b69719c98bf2a742ca8955ce7a87677c8a"}]},"branch":"refs/heads/master"},"089f475de9189d44362198afbe56cb59319dd0fc":{"kind":"REWORK","_number":6,"created":"2026-07-16 10:20:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/6"}}},"commit":{"parents":[{"commit":"b2c15ec175e173998d73f6ab20396f2337066d0b","subject":"infinidat: Simplify dependency handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b2c15ec175e173998d73f6ab20396f2337066d0b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 10:19:43.000000000","tz":540},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nBecause we don\u0027t have enough information to determine the exact lower\nbound of these libraries, the latest minor versions are used.\n\nDepends-on: https://review.opendev.org/997489\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/089f475de9189d44362198afbe56cb59319dd0fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/089f475de9189d44362198afbe56cb59319dd0fc"}]},"branch":"refs/heads/master"},"ecc663448a5cad5494c87674bce7eb8b40ac8ac2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-07-16 12:04:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/7"}}},"commit":{"parents":[{"commit":"c265e8654f286c9effe68dadde0b9a221192f0ff","subject":"infinidat: Simplify dependency handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c265e8654f286c9effe68dadde0b9a221192f0ff"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 12:04:16.000000000","tz":0},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nBecause we don\u0027t have enough information to determine the exact lower\nbound of these libraries, the latest minor versions are used.\n\nDepends-on: https://review.opendev.org/997489\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecc663448a5cad5494c87674bce7eb8b40ac8ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecc663448a5cad5494c87674bce7eb8b40ac8ac2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b2c15ec175e173998d73f6ab20396f2337066d0b","ours":"089f475de9189d44362198afbe56cb59319dd0fc","theirs":"c265e8654f286c9effe68dadde0b9a221192f0ff","merge_strategy":"recursive","contains_conflicts":false}},"1aff852ac5a82642dc93c33004f0904d91395feb":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2026-10-02 10:50:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/8"}}},"commit":{"parents":[{"commit":"b5612351dd0c58df2659d8eccba4fe8995c4bb61","subject":"infinidat: Simplify dependency handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b5612351dd0c58df2659d8eccba4fe8995c4bb61"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-10-02 10:50:29.000000000","tz":540},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nBecause we don\u0027t have enough information to determine the exact lower\nbound of these libraries, the latest minor versions are used.\n\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1aff852ac5a82642dc93c33004f0904d91395feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1aff852ac5a82642dc93c33004f0904d91395feb"}]},"branch":"refs/heads/master"},"c0bf83c7e10d615d10dfd31cb4e4772626050fce":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-10-02 10:54:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/997487/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/997487/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/997487/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/997487/9"}}},"commit":{"parents":[{"commit":"e19cdcb5c7b618c1ce7ab5cc25c06d2983858e48","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e19cdcb5c7b618c1ce7ab5cc25c06d2983858e48"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-16 08:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-10-02 10:54:57.000000000","tz":0},"subject":"Maintain driver dependencies","message":"Maintain driver dependencies\n\nSome of the share drivers require extra dependencies. Make sure that\nthese are documented and maintained.\n\nBecause we don\u0027t have enough information to determine the exact lower\nbound of these libraries, the latest minor versions are used.\n\nChange-Id: I092b58b2e3051d1f2d98451c4bfd1768ef087933\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c0bf83c7e10d615d10dfd31cb4e4772626050fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c0bf83c7e10d615d10dfd31cb4e4772626050fce"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b5612351dd0c58df2659d8eccba4fe8995c4bb61","ours":"1aff852ac5a82642dc93c33004f0904d91395feb","theirs":"e19cdcb5c7b618c1ce7ab5cc25c06d2983858e48","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
