)]}'
{"id":"openstack%2Fmanila~998567","triplet_id":"openstack%2Fmanila~stable%2F2026.1~I03eebf6bb951583378340365dbf69122f124ffc8","project":"openstack/manila","branch":"stable/2026.1","topic":"bug/2161287","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-07-24 03:57:47.000000000","reason":"Change was submitted"}},"hashtags":["ossn-0103"],"change_id":"I03eebf6bb951583378340365dbf69122f124ffc8","subject":"Prevent unauthorized resource locks search","status":"MERGED","created":"2026-07-23 22:17:11.000000000","updated":"2026-08-03 20:47:37.000000000","submitted":"2026-07-24 03:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998567-bug/2161287","cherry_pick_of_change":998388,"cherry_pick_of_patch_set":1,"meta_rev_id":"94c43e0978a55bed529fb92d1b222a00337c9ed5","_number":998567,"virtual_id_number":998567,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-24 03:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-07-24 02:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-07-24 02:16:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-23 22:39:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-23 23:10:00.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-07-24 00:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-24 00:53:04.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-07-24 02:16:24.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"ea39d2121cec0a8fd92dd007c5a2139de5956cfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-23 22:17:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2ccda2603daae9977c04d15016c57b0ed919f8cd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-23 22:39:56.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/f34e53f5e496482a9884f025276da134/","accounts_in_message":[],"_revision_number":1},{"id":"54fffbd488da7a0fdf5f3b046925710b55fa5a26","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-23 22:49:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f34e53f5e496482a9884f025276da134\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cf8040d0a871495489f40f3f895b0b53 : SUCCESS in 30m 38s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4b0905d3720f4021b9bf4b267ddf9365 : FAILURE in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3ac893a41257080bfbda7a4381d2e366ecf8372","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-07-23 23:10:00.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/998567","accounts_in_message":[],"_revision_number":1},{"id":"9306e0bc9103dc9be979997bde35da81a35befde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 00:44:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dee9302ed26a46c7acecbadbe5c60e08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2349e5fce7f34aefb101171be7f8ea63 : SUCCESS in 7m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9317d9f586084ec4b7549666c8982331 : SUCCESS in 22m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d7e4194f09452ca0fc42102eafa804 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a3a48b99b7c4e78b9968a8adae8b7a9 : SUCCESS in 7m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03a0a031ce534673bae35ae14b60f3de : SUCCESS in 7m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93e29f3518924bfaa0843fcd0358d66a : SUCCESS in 12m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff19da56d4f45cba80eb511b17ea946 : SUCCESS in 9m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8382d9013a4a4ac6a7dc7d47fa630ec5 : SUCCESS in 1h 07m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2c6097f7effe4e62a6e4b57c47d7f9a5 : SUCCESS in 2h 20m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5f493c3e45f144b4a97d92f79e361a0e : SUCCESS in 1h 18m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/fb5b5b1c491946888e8f3072bb50b63c : SUCCESS in 1h 01m 31s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2d0ec2333439418ebb3281c24154c953 : SUCCESS in 44m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b8bc84f0b62f44bbbadf4d589069725a : SUCCESS in 39m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/80ea32fa43414962ad5990114da71ed2 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37da889e151f418e9d91afed9eff6416 : SUCCESS in 4m 28s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c61212ea8dec4d09944b3fa929125f70 : SUCCESS in 35m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/65eee007e87544f5b034ca2ee300efe7 : SUCCESS in 31m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/21991aff49004d449a493cb58f972af1 : SUCCESS in 42m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f5988ef5120a4bef8ae3d4e575211030 : FAILURE in 1h 03m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/ee2969dc0a88431cabcf6cbbd28c52fb : FAILURE in 18m 19s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/23364187e1c142a39376942d29e54bd4 : SUCCESS in 1h 07m 17s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d8e6794edc442f59fdf596d660cf879 : SUCCESS in 58m 05s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a8c2e65a8a5b4fcca7ebd2e47589ee3c : SUCCESS in 38m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/87e6523dd8cf44b5a5f65dfd42a44757 : SUCCESS in 20m 21s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/aeae9397d6e04c5ebee1755c5d25bdab : FAILURE in 2m 42s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/436f7dd5c9b84360b6490e7372cf2e7a : SUCCESS in 4m 54s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/fb51c15b56474f8f9285fd153864acc5 : SUCCESS in 4m 01s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"879b710fe9d619ee7f4260d21c7100a03f8754de","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-24 00:53:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/998567/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/20063bc/ : RETRY_LIMIT in 7m 07s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f3cb608f9d188587c352e784370319925b7feefd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-24 02:15:35.000000000","message":"Topic set to bug/2161287","accounts_in_message":[],"_revision_number":1},{"id":"e8391780ea90d42cb739e3f6e025dccf818316b9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-24 02:16:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e445caa53feaac85461b8d2ba8adbe07948e1e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 02:27:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da8ff36849dc6cc0937195a7c19984d262047d03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 03:57:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef3641a213a34ad9a55f83f2f84c2a6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20de36f39c94bd081deb0f8434d9fe2 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f376e5bc86428d9abe395094026290 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/869dcf19582f493a8ac1d4fad8496e53 : SUCCESS in 12m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10f50038b37746d28c34446d1b89f25a : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60fac6a15aaa4983bb3e5885483d3044 : SUCCESS in 7m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/89a9c65454ee42e795b4cba3b037b90d : SUCCESS in 1h 18m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d2087006606e4b47b15a96c8304b4e59 : SUCCESS in 27m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0cbe9b03e223451489fa539c688bc2f3 : SUCCESS in 30m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c7501beaf0ba497c912d1ffa0265c266 : SUCCESS in 32m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f7430a321eefee5eb308404b738621593ff531aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 03:57:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c7e14ca18dc1c5f5c630466cd6e7a7b4aa9d385","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 03:58:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5b4ed4ddd440e6bc836712f681d654\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e148812d97df47d5ad20cec864dc47a2 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56dd45f638834edb821f25d97f2a2351 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"94c43e0978a55bed529fb92d1b222a00337c9ed5","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-03 20:47:37.000000000","message":"Hashtag added: ossn-0103","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8f71ea483e381f586518eefb96b10887df49848","revisions":{"a8f71ea483e381f586518eefb96b10887df49848":{"kind":"REWORK","_number":1,"created":"2026-07-23 22:17:11.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/67/998567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/67/998567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/998567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/998567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/998567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/67/998567/1"}}},"commit":{"parents":[{"commit":"9d3ab9672ab71e72c5f79b1cf3010e6f4a91994f","subject":"Merge \"Pull container image from quay.io\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9d3ab9672ab71e72c5f79b1cf3010e6f4a91994f"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-07-23 00:00:15.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2026-07-23 22:17:11.000000000","tz":0},"subject":"Prevent unauthorized resource locks search","message":"Prevent unauthorized resource locks search\n\nThe resource locks mechanism allows filtering locks by project\nUUIDs. This feature is helpful for identifying the locks and\nletting their users and/or administrators to quickly filter.\n\nHowever, in the current code path, we were allowing unauthorized\ncross-project visibility into resource locks data for users that\nwere not members, admin or readers of a given project.\n\nThis change fixes this issue by adding a check to ensure whether\nthe user has authorization to look up resource locks in the\nproject they are supplying as part of the query.\n\nCloses-Bug: #2161287\nChange-Id: I03eebf6bb951583378340365dbf69122f124ffc8\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n(cherry picked from commit 81195700d41ae216602bb508505136f2e0ac0490)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a8f71ea483e381f586518eefb96b10887df49848"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a8f71ea483e381f586518eefb96b10887df49848"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"369fe49bdbd0d5744a0f96df15a9ab4240bca570","ours":"9d3ab9672ab71e72c5f79b1cf3010e6f4a91994f","theirs":"81195700d41ae216602bb508505136f2e0ac0490","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
