)]}'
{"id":"openstack%2Fmasakari~1004321","triplet_id":"openstack%2Fmasakari~master~I4fd5f931b6bfc54ae04cab589490e4ae5ad206ea","project":"openstack/masakari","branch":"master","hashtags":[],"change_id":"I4fd5f931b6bfc54ae04cab589490e4ae5ad206ea","subject":"tox: Remove redundant injection of VIRTUAL_ENV variable","status":"MERGED","created":"2026-09-06 16:52:14.000000000","updated":"2026-09-07 00:08:46.000000000","submitted":"2026-09-07 00:07:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004321","meta_rev_id":"fb0aad065eb2a59462a29744c37988b7361a2cdb","_number":1004321,"virtual_id_number":1004321,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 00:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-06 23:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-06 23:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 17:33:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a27f0fcfa24bb1e6b500998a6daa85b9b86a327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 16:52:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16aa934c58df8ae0870b475ea56a648c06716879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 17:33:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/354b9f3d178d43839855f8b27503e398\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b0f030f6c0714c4ab9321b7a15de4a22 : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50f08110b59d435ebd9760e37a905335 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee5c15a549b4463b8537264cc0d085b : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d492b913200342d68703b47d74402032 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f95b2a9f2c6045469ccc8bceeca58160 : SUCCESS in 6m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f3ef0109f706427491fbec05639a9b80 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b9cfb12a9143fbb6ed29943b91f9c1 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6628b4c4c4f049b983a93f2a15a53a73 : SUCCESS in 3m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/02524c2891754976864ed4683c8c681f : SUCCESS in 4m 05s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/d370bbb63e1d44778b0b03821a8c2bde : SUCCESS in 20m 10s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/b953ff166bb146f4801ea61406ca02b3 : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3b636998f590c728607d82cbd5a9e169951906c8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 23:21:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"145fb8343c94ff57ad721429ee232783ff15e527","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 23:22:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1aee5c9445b745b38ac619e6a44e5e814d4feff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 00:07:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c34a11cdc934e7f9f1485e2d6d64ba3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3100fbbecf1f4f718976820d18ef74a9 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5219b7d1c644c0aa7cec083edde7c53 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a887aca181414228bd0be27b8f8972fd : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53de64b5df82433fa5b0aebe187b6259 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8ed5a86f7442d7a98188ef9b76a5b3 : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64730efeeb204bfe8c5f782e3968a383 : SUCCESS in 4m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5626a91980614decab38f70e4c9e00e8 : SUCCESS in 5m 18s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/491a85cf0128411fbb2dd7e959f64877 : SUCCESS in 39m 56s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/75da184735584fa6965b9c23adf6ca01 : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c5c622ea4746af15935fef9fa58b0ac49014cb64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 00:07:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fb0aad065eb2a59462a29744c37988b7361a2cdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 00:08:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f701b167b4d1471fbfa5adf3ad6168aa\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8344dcc2bfdb4671b053d9c7ee3a5b0f : SUCCESS in 54s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e063dcaa7bda452cbe0579367f3c825b : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/209f3543d163468f9f0466c12826ba0f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba5bbab2bbd255dbecc563e39989c18440526072","revisions":{"ba5bbab2bbd255dbecc563e39989c18440526072":{"kind":"REWORK","_number":1,"created":"2026-09-06 16:52:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/1004321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/21/1004321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/1004321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/1004321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/1004321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/21/1004321/1"}}},"commit":{"parents":[{"commit":"e156554c671ed731502cb17c48feb0f3467e712a","subject":"tox: Drop redundant passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/e156554c671ed731502cb17c48feb0f3467e712a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 16:51:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 16:52:07.000000000","tz":540},"subject":"tox: Remove redundant injection of VIRTUAL_ENV variable","message":"tox: Remove redundant injection of VIRTUAL_ENV variable\n\nThe same is already set by tox[1].\n\nTrivial-Fix\n\n[1] https://tox.wiki/en/latest/reference/config.html#injected-environment-variables\n\nChange-Id: I4fd5f931b6bfc54ae04cab589490e4ae5ad206ea\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/ba5bbab2bbd255dbecc563e39989c18440526072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/ba5bbab2bbd255dbecc563e39989c18440526072"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
