)]}'
{"id":"openstack%2Fmasakari~670710","triplet_id":"openstack%2Fmasakari~master~I8b3936949dda2029901e5f761d7b1a1e0e4fc9a4","project":"openstack/masakari","branch":"master","hashtags":[],"change_id":"I8b3936949dda2029901e5f761d7b1a1e0e4fc9a4","subject":"Add Python 3 Train unit tests","status":"ABANDONED","created":"2019-07-14 06:48:47.000000000","updated":"2019-09-20 03:29:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee525d71c6a125864245685ce708e0f6a01ba93f","_number":670710,"virtual_id_number":670710,"owner":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-14 10:24:39.000000000","_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-14 11:14:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-14 10:24:39.000000000","updated_by":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-14 11:14:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e93a745872da6935e376347596c459b46797559e","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-07-14 06:48:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9587867c6d1da0e0a94f069874e31fe28709d06d","author":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-14 06:49:10.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/masakari on branch master.  The error was:\n\n  The project template \"openstack-python3-train-jobs\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - masakari-functional-devstack-multinode:\n              voting: True\n\n  in \"openstack/masakari/.zuul.yaml@master\", line 81, column 3","accounts_in_message":[],"_revision_number":1},{"id":"9eee56766123ec96fea29e9cc8385889a219a625","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 07:45:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/10/670710/1/check/test-release-openstack/5144643/ : SUCCESS in 2m 14s\n- build-openstack-api-ref http://logs.openstack.org/10/670710/1/check/build-openstack-api-ref/ef6e345/html/ : SUCCESS in 2m 23s\n- openstack-tox-cover http://logs.openstack.org/10/670710/1/check/openstack-tox-cover/a06fb06/cover/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/670710/1/check/openstack-tox-lower-constraints/ab52827/ : SUCCESS in 2m 42s\n- openstack-tox-pep8 http://logs.openstack.org/10/670710/1/check/openstack-tox-pep8/d9f57f0/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/10/670710/1/check/openstack-tox-py27/8df7795/ : FAILURE in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/10/670710/1/check/openstack-tox-py36/5f829c4/ : SUCCESS in 3m 06s\n- openstack-tox-py37 http://logs.openstack.org/10/670710/1/check/openstack-tox-py37/ebb5ac1/ : FAILURE in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/10/670710/1/check/openstack-tox-docs/157349c/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/10/670710/1/check/build-openstack-releasenotes/b1b0b1e/html/ : SUCCESS in 2m 31s\n- masakari-functional-devstack-multinode http://logs.openstack.org/10/670710/1/check/masakari-functional-devstack-multinode/d51fa40/ : FAILURE in 55m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7b61c41531d58c51d3e2f5f44b21b6cd3feba09b","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-07-14 10:24:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17e62a125961775aa7beae5985b632abe9cd82e3","author":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-14 10:24:39.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/masakari on branch master.  The error was:\n\n  The project template \"openstack-python3-train-jobs\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - masakari-functional-devstack-multinode:\n              voting: True\n\n  in \"openstack/masakari/.zuul.yaml@master\", line 81, column 3","accounts_in_message":[],"_revision_number":2},{"id":"11df03336baca9e07b208e57acf11ec48f66eba6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 11:14:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/10/670710/2/check/test-release-openstack/28b984e/ : SUCCESS in 2m 04s\n- build-openstack-api-ref http://logs.openstack.org/10/670710/2/check/build-openstack-api-ref/9ce3984/html/ : SUCCESS in 2m 59s\n- openstack-tox-cover http://logs.openstack.org/10/670710/2/check/openstack-tox-cover/f576e55/cover/ : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/670710/2/check/openstack-tox-lower-constraints/3f7a6ad/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/10/670710/2/check/openstack-tox-pep8/8366bb9/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/10/670710/2/check/openstack-tox-py27/3ea4c13/ : FAILURE in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/10/670710/2/check/openstack-tox-py36/16ca72a/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/10/670710/2/check/openstack-tox-py37/4f111c2/ : FAILURE in 2m 33s\n- openstack-tox-docs http://logs.openstack.org/10/670710/2/check/openstack-tox-docs/e0d0a2a/html/ : SUCCESS in 2m 52s\n- build-openstack-releasenotes http://logs.openstack.org/10/670710/2/check/build-openstack-releasenotes/2e28853/html/ : SUCCESS in 2m 51s\n- masakari-functional-devstack-multinode http://logs.openstack.org/10/670710/2/check/masakari-functional-devstack-multinode/5a7d290/ : FAILURE in 44m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ee525d71c6a125864245685ce708e0f6a01ba93f","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-09-20 03:29:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a709e302f194b0ff29960177f769d80fc0919b35","revisions":{"3fc337db62fb934f5d87dae84f82358247a07d39":{"kind":"REWORK","_number":1,"created":"2019-07-14 06:48:47.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/10/670710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/10/670710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/10/670710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/10/670710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/10/670710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/10/670710/1"}}},"commit":{"parents":[{"commit":"e8a18eee8314388c949e4f1a3c4a62b6cfe0d2b2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/e8a18eee8314388c949e4f1a3c4a62b6cfe0d2b2"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-07-14 05:43:32.000000000","tz":0},"committer":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-07-14 06:48:29.000000000","tz":0},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I8b3936949dda2029901e5f761d7b1a1e0e4fc9a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/3fc337db62fb934f5d87dae84f82358247a07d39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/3fc337db62fb934f5d87dae84f82358247a07d39"}]},"branch":"refs/heads/master"},"a709e302f194b0ff29960177f769d80fc0919b35":{"kind":"REWORK","_number":2,"created":"2019-07-14 10:24:16.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/10/670710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/10/670710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/10/670710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/10/670710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/10/670710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/10/670710/2"}}},"commit":{"parents":[{"commit":"e8a18eee8314388c949e4f1a3c4a62b6cfe0d2b2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/e8a18eee8314388c949e4f1a3c4a62b6cfe0d2b2"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-07-14 05:43:32.000000000","tz":0},"committer":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-07-14 10:22:01.000000000","tz":0},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I8b3936949dda2029901e5f761d7b1a1e0e4fc9a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/a709e302f194b0ff29960177f769d80fc0919b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/a709e302f194b0ff29960177f769d80fc0919b35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
