)]}'
{"id":"openstack%2Fmasakari~751007","triplet_id":"openstack%2Fmasakari~master~I7d8f1d254f1ce4f568efd273c39fa01170505d16","project":"openstack/masakari","branch":"master","hashtags":[],"change_id":"I7d8f1d254f1ce4f568efd273c39fa01170505d16","subject":"Fix l-c testing for Ubuntu Focal (py38)","status":"ABANDONED","created":"2020-09-10 14:37:31.000000000","updated":"2020-09-11 17:32:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cdcc14d084edd0ddbdc7133395abf2aad7f6a390","_number":751007,"virtual_id_number":751007,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-10 16:00:36.000000000","_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-10 15:15:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 15:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-10 16:00:36.000000000","updated_by":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ec4094e13d85dc9cb5246fc6c1eea2ee2326527","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-10 14:37:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89de749f93ea9fe90ce1e4e9881fcbde3856eb08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 15:15:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf9f3d07fe944ca2b8b6784a7701169a : FAILURE in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3911229bb1f42fe808ad2859024d658 : FAILURE in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aca69745a3c54f1f820bb674bdb3ca20 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058dee35d8cc4d5d9f85b794f8bb2e74 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcd974ec8fd34d3086a7308dc723c030 : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5820d27494b84d628aa41bed153a4ef8 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01652dafd024a99b79cd49ad972de8b : SUCCESS in 4m 49s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/cdf5c08a367b41948580937205fbad3b : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":1},{"id":"937e2c4cf95dfe6454a1bebf0283014c0a0bdc07","author":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-10 16:00:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. Contact info: masakari.integration.test@gmail.com . For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/NTT_SystemFault_MasakariIntegration_CI\n\n- masakari-build-test http://153.156.76.189:8000/751007/1/20200910233825 : SUCCESS in 1h 22m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cdcc14d084edd0ddbdc7133395abf2aad7f6a390","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-11 17:32:49.000000000","message":"Abandoned\n\nhandled in https://review.opendev.org/751271","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd8da2e6ce7723d82bc931a3b8702191aa33f786","revisions":{"dd8da2e6ce7723d82bc931a3b8702191aa33f786":{"kind":"REWORK","_number":1,"created":"2020-09-10 14:37:31.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/07/751007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/07/751007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/07/751007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/07/751007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/07/751007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/07/751007/1"}}},"commit":{"parents":[{"commit":"112610abce26e90abb393b1374d4ebc295275a7d","subject":"remove future from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/112610abce26e90abb393b1374d4ebc295275a7d"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-10 14:37:05.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-10 14:37:05.000000000","tz":480},"subject":"Fix l-c testing for Ubuntu Focal (py38)","message":"Fix l-c testing for Ubuntu Focal (py38)\n\nChange-Id: I7d8f1d254f1ce4f568efd273c39fa01170505d16\nStory: 2007865\nTask: 40199\nCloses-Bug: #1886298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/dd8da2e6ce7723d82bc931a3b8702191aa33f786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/dd8da2e6ce7723d82bc931a3b8702191aa33f786"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
