)]}'
{"id":"openstack%2Fmasakari~754752","triplet_id":"openstack%2Fmasakari~master~I218cadae63065fb650e1a52ea9ced3c1a27441ed","project":"openstack/masakari","branch":"master","topic":"reno-victoria","hashtags":[],"change_id":"I218cadae63065fb650e1a52ea9ced3c1a27441ed","subject":"Update master for stable/victoria","status":"MERGED","created":"2020-09-28 15:43:36.000000000","updated":"2020-10-14 16:40:18.000000000","submitted":"2020-10-14 16:38:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754752-1602693526559-11ac1cc1","meta_rev_id":"b8f7dc05e20c252306b283f9658920e0654fb167","_number":754752,"virtual_id_number":754752,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-28 16:10:15.000000000","_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 16:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-10 22:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-09-29 11:24:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2020-10-14 16:29:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-28 16:10:15.000000000","updated_by":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-10 22:23:44.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-10-14 16:29:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-14 16:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"775a940efb916ec8d0670490d09efae557d347c3","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-28 15:43:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"170543ef193db832ca4e094b66ca2af7d496161f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 16:05:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/144405453344481a8d358691ade19fbb : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ddb323e7a9420c8af8a0669f5b0765 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e7653e9e66347b7941ab2bacd23e8b7 : SUCCESS in 4m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ed0027238ce94675adcf0fb9492dd0c0 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"131f372bd2808e1c2ab9a0e5219478cfb1ab9f2c","author":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-28 16:10:15.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-masakari-systemfault-integration-ci\u0027 Contact info: masakari.integration.test@gmail.com . For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/NTT_SystemFault_MasakariIntegration_CI\n\n- masakari-build-test http://153.156.76.189:8000/754752/1/20200929004459 : FAILURE in 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"718e46e7feb67728170a5727df2baa16546358a1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-29 11:24:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d65b88b35ebbc0e9f7f0a95cbebbe4c4c4384e5a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-10 22:23:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d4a738969a867c3fc8778dfc634749f38ca016c2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-14 16:29:51.000000000","message":"Patch Set 1: Workflow+1\n\nwe got released so let\u0027s merge already","accounts_in_message":[],"_revision_number":1},{"id":"9e932ddb6c332aeff6537ad37a67e0a4903c38ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:30:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c3bf61419eb671fa77ad85d1d08393b545e570f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:38:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d9ddb0b2c9a823a0ad7c9d72ccab969a0e55dfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:38:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3d56e49fa04900a6a712cfbb185da5 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6fe7c42adb459683e4b7ffe8738a3e : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25cb003ed7ec4de8af7d4d7219b004a5 : SUCCESS in 3m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/117e54a109e242fea2bdc99f1ad8e62f : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e8f67bc5d6becc18fca987777e633b18485ee3bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:40:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93eb9900689c432181a8511b8493bc51 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a0cebf33fc143bea11f51fbba35b70e : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30fcf89b755d3dfadfa038190279fbda6139f9e1","revisions":{"30fcf89b755d3dfadfa038190279fbda6139f9e1":{"kind":"REWORK","_number":1,"created":"2020-09-28 15:43:36.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/52/754752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/52/754752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/52/754752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/52/754752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/52/754752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/52/754752/1"}}},"commit":{"parents":[{"commit":"3fcadf85c73bbd0ca1be6a359369ac3a18fc1d71","subject":"Merge \"check expired notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/3fcadf85c73bbd0ca1be6a359369ac3a18fc1d71"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 15:43:34.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 15:43:34.000000000","tz":0},"subject":"Update master for stable/victoria","message":"Update master for stable/victoria\n\nAdd file to the reno documentation build to show release notes for\nstable/victoria.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/victoria.\n\nChange-Id: I218cadae63065fb650e1a52ea9ced3c1a27441ed\nSem-Ver: feature\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/30fcf89b755d3dfadfa038190279fbda6139f9e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/30fcf89b755d3dfadfa038190279fbda6139f9e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
