)]}'
{"id":"openstack%2Fmasakari~754753","triplet_id":"openstack%2Fmasakari~master~I87829c220c4a9e8664bc152b97b36e5d2899543c","project":"openstack/masakari","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"I87829c220c4a9e8664bc152b97b36e5d2899543c","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-28 15:43:39.000000000","updated":"2020-10-14 17:00:22.000000000","submitted":"2020-10-14 16:48:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754753-1602694139310-75ebf145","meta_rev_id":"d28f7fe8ef9164bd581ed30534fd402f61cdbd9c","_number":754753,"virtual_id_number":754753,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-28 15:44:26.000000000","_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 16:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-29 11:26:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-14 16:32:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-28 15:44:26.000000000","updated_by":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 16:32:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-14 16:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c001cb0f99dcbcdd7d6cd559ebd0a5ac3939de35","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-28 15:43:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"373d384672b19eea5bc2854b4b8767d8262e29c7","author":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-28 15:44:26.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/masakari on branch master.  The error was:\n\n  The project template \"openstack-python3-wallaby-jobs\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python3-wallaby-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-linters\n          - masakari-functional-devstack-multinode:\n              voting: true\n      gate:\n        jobs:\n          - openstack-tox-linters\n\n  in \"openstack/masakari/.zuul.yaml@master\", line 136, column 3","accounts_in_message":[],"_revision_number":1},{"id":"fc0d8118683752250cdefa7b47ee833f0802d31e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 16:14:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f38369da4b44c898427d65b2116bea7 : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64613261734a4d4bbb5bb4c7340af6aa : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a3b53066d14fea9260c8782c6c8828 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64d2ea94b4a74ea99b7cd40ab86034ba : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/236cb2c5e8124f4ba8120de727ec0bff : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce41e46e56c94c3bba3a6a0b9af68cd3 : SUCCESS in 7m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/89d05a65c37d4b71933ae39b63a6511d : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"195dfe2ea2cbaf14c856db29db2639adf14d4d9f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-29 11:26:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"21f418936c1a840277e706558102eaa143e0a95f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-14 16:32:58.000000000","message":"Patch Set 1: Workflow+1\n\nwe got released so let\u0027s merge already","accounts_in_message":[],"_revision_number":1},{"id":"580e8e25172bc5c149840e93405f2de5fb0e5fdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:33:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8574420c1650dea8751c4512fff53432bab8cae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:48:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"879a2f4870fd565d23d4d91e11e973ea6b39003e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:48:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db324ba77c3242be83d720ae6d38b588 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d04ffdb19a8d4054abdc8343547353ee : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/126ba74ac99b4aac89213e7b17e8e542 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0efd6c80c103455783d435995eeb745a : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b1218b5d7648f7be729a0b460e64ca : SUCCESS in 5m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d13d3d096b134dbf8e0eb018fa56074e : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f41ae786986718c3af0aa33318389f41e6177061","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 17:00:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e1a071ca3b43da9d6e7a5527023016 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d3791d12ab77ae979b73ba94784fbd5d7c5fcc9","revisions":{"6d3791d12ab77ae979b73ba94784fbd5d7c5fcc9":{"kind":"REWORK","_number":1,"created":"2020-09-28 15:43:39.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/53/754753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/53/754753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/53/754753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/53/754753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/53/754753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/53/754753/1"}}},"commit":{"parents":[{"commit":"30fcf89b755d3dfadfa038190279fbda6139f9e1","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/30fcf89b755d3dfadfa038190279fbda6139f9e1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 15:43:36.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 15:43:36.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I87829c220c4a9e8664bc152b97b36e5d2899543c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/6d3791d12ab77ae979b73ba94784fbd5d7c5fcc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/6d3791d12ab77ae979b73ba94784fbd5d7c5fcc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
