)]}'
{"id":"openstack%2Fmasakari~874049","triplet_id":"openstack%2Fmasakari~master~I10b4793d0471a26df4381988670334a695067d27","project":"openstack/masakari","branch":"master","hashtags":[],"change_id":"I10b4793d0471a26df4381988670334a695067d27","subject":"vmoves table purge support","status":"MERGED","created":"2023-02-16 03:12:13.000000000","updated":"2023-02-16 06:38:54.000000000","submitted":"2023-02-16 06:37:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874049","meta_rev_id":"3cdd3550f8733a1383e5bcfadafb57757cd31ec7","_number":874049,"virtual_id_number":874049,"owner":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 06:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-16 05:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-16 05:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 03:50:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0effcd2dd2d4739a062a661bb109d760f705f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2023-02-16 03:12:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0927edda5870dee10fdd279dbfffc997a179e69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:50:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a630951b979d45d5bbad8f237669a9de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67a73fc470eb4c6d8b357569db085edc : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43dda9e634c04a0cbd726852ef2498f6 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/663ce7f5b24349d2be74f222a2ecc92f : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b9049d02ba14dea88ce41ac9ddd9a5b : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a32acb7097534cddb5a85a70fc8067f6 : SUCCESS in 6m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/592efefc48c04553975377cd20e4ed84 : SUCCESS in 3m 38s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/f41270f5ba354798bedde84a9adba374 : SUCCESS in 31m 40s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/95b9173712ec4de3a03c77a1028526e4 : SUCCESS in 31m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a24d12053504658bebff0c9171178cb8099489de","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2023-02-16 05:40:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6136cb3660ee3e32697bbf20417372f3395408aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 05:41:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b8075049729282d62aafe968b9b614ffb76884f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:37:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd0b16f003dc4335ab7f2d84efac85d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2551861b221d4e55b167f3400f0f718b : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8203254def3479d9a75a9f04a24b661 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ea2d8bf104d4a70b2a80629e2a06c36 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e573eba95c745d6b62509bd45d2c96b : SUCCESS in 7m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cf269ee766a046f2b95ea602f26b98a5 : SUCCESS in 3m 50s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/ea6cbe4f043144ee848de2f98b7fc9c7 : SUCCESS in 41m 45s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/c1c4a5d26dee419bbacc7d92d700ecb2 : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7888d6671203e050d7762efd442435637e0bd598","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:37:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3cdd3550f8733a1383e5bcfadafb57757cd31ec7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:38:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a9864279164839834e04a2fc96b093\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d807fbc9cde648999ed698e21b0f6a1b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e1b85133a59ee68a3b18e47e7dc49ab30097e63","revisions":{"3e1b85133a59ee68a3b18e47e7dc49ab30097e63":{"kind":"REWORK","_number":1,"created":"2023-02-16 03:12:13.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/49/874049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/49/874049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/49/874049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/49/874049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/49/874049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/49/874049/1"}}},"commit":{"parents":[{"commit":"20db09fe0a4f3759cdfda8b180abfe2ca5761b06","subject":"Merge \"add api-ref for masakari vmoves\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/20db09fe0a4f3759cdfda8b180abfe2ca5761b06"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-02-16 03:11:01.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-02-16 03:11:01.000000000","tz":480},"subject":"vmoves table purge support","message":"vmoves table purge support\n\nChange-Id: I10b4793d0471a26df4381988670334a695067d27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/3e1b85133a59ee68a3b18e47e7dc49ab30097e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/3e1b85133a59ee68a3b18e47e7dc49ab30097e63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}}]}],"submit_requirements":[]}
