)]}'
{"id":"openstack%2Fmasakari~949721","triplet_id":"openstack%2Fmasakari~master~Iae4aec9350e071bdec592683d8e0a9f4d1823146","project":"openstack/masakari","branch":"master","attention_set":{},"removed_from_attention_set":{"30623":{"account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"last_update":"2025-06-03 01:04:41.000000000","reason":"\u003cGERRIT_ACCOUNT_30623\u003e replied on the change","reason_account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}}},"hashtags":[],"change_id":"Iae4aec9350e071bdec592683d8e0a9f4d1823146","subject":"HA notification","status":"MERGED","created":"2025-05-14 10:30:17.000000000","updated":"2025-06-03 01:52:22.000000000","submitted":"2025-06-03 01:51:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949721","meta_rev_id":"d8184e6a974662c434f9ffc97bf543498d6530a7","_number":949721,"virtual_id_number":949721,"owner":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 01:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-03 01:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-03 01:04:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-14 11:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-31 21:07:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"f3fd56015fb19176b06a95cdf83b70a9c02941da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-05-14 10:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ab6c2a57e47313fbe1fed53dd3d637687c2f875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 11:08:37.000000000","message":"Patch Set 1: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/222825ff07c1489d90318c912b8c3fa7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd39d3fb287f425ea792bd66857757af : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2781cb9139834a3495fc4ee75fae1a7d : FAILURE in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/841e13e7a3304a57917749f19e0ede42 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6aeed11a93724e71b2cbaa4ff889078c : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51b9de85076b4251a1943fb04e947aab : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3ae80c979a46d6a8448482f792d2aa : SUCCESS in 4m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/008ac167b03045179b8c07da9f4ec383 : FAILURE in 2m 49s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/a06b9cba5e394230a04605f23e78ab26 : FAILURE in 18m 41s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/d36645335bf04f74a0e18c2268366bc2 : FAILURE in 36m 26s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/161332a4d1ba46338d6c3cd27660a973 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1fe9b89346249dd785357db3b8e64363861e4c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-05-15 02:43:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae7134c0ed28fa6119657975f3ece90ddf6b568c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:21:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dc6ceb4d8ce4b8f82712acc20571597\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3f0326d52f24452b6bfd39883064432 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad156d0e27146b6948cf6d693ae8190 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0787d4974fdc4e16b09ddd7ef89e8331 : SUCCESS in 2m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1b86e2236144377a0e292737692d338 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/967308a5ae894594bd8f761ad529cc2d : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150db4c00ade4f038e8b007c4984099d : SUCCESS in 7m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0fdd0f26b1c04e7489318877453a3de6 : SUCCESS in 4m 03s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/feff8c173891457aaf586ca54d5793ff : FAILURE in 33m 10s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/3b1edeadcd4c4b6ea7f7be445e531405 : FAILURE in 28m 34s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/66ab51e22c47480cb6fa3f886db665f7 : FAILURE in 24m 58s","accounts_in_message":[],"_revision_number":2},{"id":"76e5d8f14c26d857dd2ae8be7676ff864ba87967","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-31 21:07:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d262116e415d24c1adc2b02b88ae685acdfe7928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 21:54:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2abbdb321347a4b3686cd18c507769\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/980c205a35db41b5b6c8d279ef56cfac : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81e41ac58c04411b8e3056c0d2c2fe1 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a477c2b732e84e13bbf2fbe159db19d8 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/deab149368ae4015a91d17f8a18f6183 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d803c040126455cb7e9d409b9428200 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d537232431b74466baee13feeeee51db : SUCCESS in 6m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e91486acdf164e7189c00aa89bc1a04a : SUCCESS in 3m 20s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/30a5b315a1034c51a6db0570e9cfcae8 : SUCCESS in 24m 43s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/8c20f09faf414300a49208f3f6d37859 : SUCCESS in 26m 41s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/767d9d67a01e47b0a59bd51a845734c0 : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":2},{"id":"c325d0c604e6eb152fc909fdd21c5c45d8e48db2","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-06-03 01:04:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6e7723c32cab381a20c1fdcee23b571fb38903bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 01:05:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8deddc576c0388d9301a2e56273882b84ec83244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 01:51:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66c3ea65c14b4aed8b1fec7621b33999\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ad794b4cb342908cd65f7a5da4a12a : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7339776df50490ab9f076594874d488 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05e287f4247b435fadbf0bed604b9215 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055a5e5423a44e25b053f244f5d68648 : SUCCESS in 9m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af8027000c424510a0e7de1fdb549b56 : SUCCESS in 3m 56s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/5e058942c7354b3ebd1149bb65e44954 : SUCCESS in 34m 40s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/8061e070643b499c94a97e04898a6343 : SUCCESS in 36m 22s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/f7cec5c030e84be48c49a54cbd705647 : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":2},{"id":"de4d0d033096960d826fdf64cdbcc3ab788a805d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 01:51:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d8184e6a974662c434f9ffc97bf543498d6530a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 01:52:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fca41ced8fa413f851f2cff3aa3d470\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f9f3ff74ed41659fc66ca69b732fff : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a49b18f3a242e2186fffcc4ac03094b95cd3d41d","revisions":{"cb049c0f5ebf8e3245e306aae0ac3cd84cf91a85":{"kind":"REWORK","_number":1,"created":"2025-05-14 10:30:17.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/21/949721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/21/949721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/949721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/949721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/949721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/21/949721/1"}}},"commit":{"parents":[{"commit":"01cf6692f3dfa97600d8bc02bb34559526d29646","subject":"Merge \"Update master for stable/2025.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/01cf6692f3dfa97600d8bc02bb34559526d29646"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-05-14 10:29:21.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-05-14 10:29:29.000000000","tz":480},"subject":"HA notification","message":"HA notification\n\nIt adds failover_segment_uuid and message in notification object.\nThen the user can filter all the notifications by failover_segment_uuid.\nThe user also can get some information of the notification by message\nfiled.\n\nChange-Id: Iae4aec9350e071bdec592683d8e0a9f4d1823146\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/cb049c0f5ebf8e3245e306aae0ac3cd84cf91a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/cb049c0f5ebf8e3245e306aae0ac3cd84cf91a85"}]},"branch":"refs/heads/master"},"a49b18f3a242e2186fffcc4ac03094b95cd3d41d":{"kind":"REWORK","_number":2,"created":"2025-05-15 02:43:27.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/21/949721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/21/949721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/949721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/949721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/21/949721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/21/949721/2"}}},"commit":{"parents":[{"commit":"01cf6692f3dfa97600d8bc02bb34559526d29646","subject":"Merge \"Update master for stable/2025.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/01cf6692f3dfa97600d8bc02bb34559526d29646"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-05-14 10:29:21.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-05-15 02:43:10.000000000","tz":480},"subject":"HA notification","message":"HA notification\n\nIt adds failover_segment_uuid and message in notification object.\nThen the user can filter all the notifications by failover_segment_uuid.\nThe user also can get some information of the notification by message\nfiled.\n\nChange-Id: Iae4aec9350e071bdec592683d8e0a9f4d1823146\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/a49b18f3a242e2186fffcc4ac03094b95cd3d41d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/a49b18f3a242e2186fffcc4ac03094b95cd3d41d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
