)]}'
{"id":"openstack%2Fmasakari~951613","triplet_id":"openstack%2Fmasakari~master~I30f3df8dc889b64679884944084579815f732314","project":"openstack/masakari","branch":"master","topic":"large-scale-failure","attention_set":{"30623":{"account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"last_update":"2025-06-03 07:55:06.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I30f3df8dc889b64679884944084579815f732314","subject":"evaluator rules for ha workflow","status":"NEW","created":"2025-06-03 07:12:57.000000000","updated":"2025-08-19 13:05:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca59eb819183dc5d65da527a834ceb8f6e738608","_number":951613,"virtual_id_number":951613,"owner":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-19 13:05:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 07:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 20:46:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"df09ef37fdc2309bb852bfbe82e6720f73ff15dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-06-03 07:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05219406efdb416de6092655b55d4d7240b99881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 07:55:06.000000000","message":"Patch Set 1: Verified-1\n\n(34 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e20f9b510e8d4e168f59bdd084aabebe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6bb387631dbe48e4b762b6a40c0eb3d3 : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e6ab675a5684a9f94630a1b4b1ea0c7 : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a49a9cb464274e4c8139da2bc79bb27d : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e16474f07a64b35bb6b87ad4de4b447 : FAILURE in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/339a385840bc4ec78f8d7aaad91883ca : FAILURE in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ef73c0af82445c5bdc3cb590da18589 : FAILURE in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6abc2c3ac946d2bccede39adcc5c20 : SUCCESS in 7m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bb9075a0284d4548bdb31d2866f9ee0f : FAILURE in 4m 00s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/70d7dde080f345c4a73728b519867021 : FAILURE in 39m 45s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/331519ea90d442219e975c3219889b8e : FAILURE in 40m 28s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/e946bed24f5f44ddaaa7fbedc7f15e57 : FAILURE in 39m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2d9c3477ccfcb18f598d0ac50d1d74109875f8f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-06-11 02:28:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d554b47a72787921b821c52c5f4e9fede115af9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 03:27:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfb0a8e9d47c4831bee9cc1f611799f8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e9728249a984e25a08124a5819b3ed2 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8137667e94234084b2e979f136f63272 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d970ab6bdb646aa99b5f72afc19295b : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1aac45f69794cb795131f429fee99dc : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f85c1c8a15744ac8e4e0acff7fcb114 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64095c7637d541579e808506ee58e135 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/281acaaa699d49b7ab22373438957fe5 : SUCCESS in 6m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5b39638256154a9f8a506e87d0016a87 : SUCCESS in 3m 41s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/2b7708ee40e84b09a8b1994b3058fcb9 : FAILURE in 45m 57s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/895d4feecb594383a6fc738aa5e122b6 : FAILURE in 45m 54s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/8de707f315984e3f996c72c46da2ac6a : FAILURE in 25m 53s","accounts_in_message":[],"_revision_number":2},{"id":"27440aec5efebb65800c1a5b208c5c542dea1643","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-20 20:46:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69e160f0d84c20eef8027114c356091ccae3debd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 21:42:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8dc0fad30ab4fa0acdd576a3dd35f6a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b622e43303834b98aed1b7dff0e8c075 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb864a73bfaa4bf6bba4e1f2476f84de : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb8ee4c9a264169857fe718136bac07 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c528a24b324949308128a9e7227a0db0 : SUCCESS in 6m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/729be4904a7b47ae9efc40ce7b8f62c0 : SUCCESS in 6m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fe198a2cf5f4f6ea9bdda1030f5cbb9 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd521124b1fe45a4affbfb88c18d1418 : SUCCESS in 9m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5f043ff6fe4346d28f7078fe31a9d7fe : SUCCESS in 3m 36s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/f20ae34f6b6c46c184abb9de47be64dc : FAILURE in 45m 33s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/c645896a0e7a48498a5f0a63a5752630 : FAILURE in 46m 06s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/3fbeabc8f033434e8093bb5bc2f5b901 : FAILURE in 34m 40s","accounts_in_message":[],"_revision_number":2},{"id":"9accae5a1410e9fcfbe37aad1d9faf5789f77e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-06-27 08:01:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"12c2f9726a22d8614ed89c1ca4cf92e45ad910eb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-06-27 08:02:04.000000000","message":"Topic set to large-scale-failure","accounts_in_message":[],"_revision_number":3},{"id":"5e203691e31645a669c5246a525be3e21ac74569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 08:45:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ef0fdbe0da04455b827caffcabe162a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e1b52f440244c839a74a459ec0288bd : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/044967f28d1d47ff8d17de8764142882 : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e18978c76eec4735afafa886ac6395c0 : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12927112e71f47e2bc38f8ea18d3bf47 : FAILURE in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef7e8d8b966149bcb267992d3859fe2c : FAILURE in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fe99d342c4744dcab70ce28c1914675 : FAILURE in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0896b50027ed4995a569420d71f9145b : SUCCESS in 7m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ed382cb8dae3424089210858605f0d20 : SUCCESS in 4m 14s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/27846b1d3bdc440fb039bc6121646fe9 : FAILURE in 30m 00s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/071279a42cdf42ae9d9ad08a766db6d2 : FAILURE in 43m 54s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/55a287f8a2cf400eadcc4122c961bb98 : FAILURE in 21m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9afcff652afa68bccdfa86872526aa858cd4fb5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-06-27 09:30:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f23c1c34ffde6141048c74284b4c4f128ab3edaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:12:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0ab15bcb57c48ea800fd71030838148\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fedf5b9ce53640d480fc3c5f67f188ea : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/784b3009468a441e8f8f4823ab5d68ec : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b48c262f0ac4bfbad41e8e4e27c0caa : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2c6c8cf09ef4df1bb9cd86b40f7afa9 : SUCCESS in 5m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b00b51a00634cc7a7eb552490a937a6 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a9c9f220154448f8081ec50dd186a54 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72a0f5944964e9b97e7bf80046e6134 : SUCCESS in 4m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fd95c3bdb9664e88a4e49e3f28e3a602 : SUCCESS in 3m 33s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/426b590a8a4b43dab7d645004b4f88df : FAILURE in 39m 50s\n- masakari-functional-devstack-multinode-jammy https://zuul.opendev.org/t/openstack/build/241a443d8f6a49ffba794b9f63ebc01a : FAILURE in 25m 58s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/8d2ba9adcd5a4bc195553e1572bf6934 : FAILURE in 25m 14s","accounts_in_message":[],"_revision_number":4},{"id":"adf1d5c93b3bdbb9e7618cac61ae5ba263d890aa","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-13 09:38:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"337e1057ef2dc1726856aea63878c2e549d0d494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 10:22:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5abfd93c76d2498bac583635f3abe6e9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a96287e32ff1433abe22c504a8d5251c : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc543072562e484a8f91589c60c2d59e : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759ef81e64184687b9a060b1ef734fd5 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/433fe78dbfb945ba8b7f383b883a8225 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88cd706c552f44458bad873a0cf39761 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d216f3f8cf84bae8bfbcfcdb86adbfe : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd571eab1f4a4c86afae37c7027925eb : SUCCESS in 7m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/951af88e5e3c4467b682934c86e1f5d5 : SUCCESS in 3m 33s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/1e24d6f87c394a85baa5bc1aab5aafce : FAILURE in 41m 59s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/086bd61aed5447feae82ed7a92611651 : FAILURE in 21m 48s","accounts_in_message":[],"_revision_number":4},{"id":"75d58b9e8e564461101f8f32d4372717809e12fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-08-19 09:09:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"647fc35dcf0e44f0743aa92154a4c43993dcb4e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 09:27:07.000000000","message":"Patch Set 5: Verified-1\n\n(18 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5569b46466574b178c9297bcc69c2ffd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/47d58c6215d14ee0910160f0e2d8407b : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a0e769099de44eb9b0b2d5975ac8cb1 : FAILURE in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/625f9123f6974cad99fb591775eccc50 : FAILURE in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/110e5af6f7f143e89c885cd138160f74 : FAILURE in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55f7ba8d1e834ef68979f3ed2dc78ecc : FAILURE in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e438755593a84f40b8744198b2def771 : FAILURE in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f0546129104b1c841b9ab2b35766a6 : SUCCESS in 7m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5058b57d25554c9f84251e108b6401c3 : FAILURE in 4m 08s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/e2dcba531fe14412919b7acb78fa175a : POST_FAILURE in 15m 28s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/a67bdcd7c6034b63870fd48402981bac : FAILURE in 13m 27s","accounts_in_message":[],"_revision_number":5},{"id":"5605f74724dc88d1f299cb22ca3ecf9f8310fe6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-08-19 11:12:22.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"834996ba8dd63bd3e82ac37dd1604facbfee0794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:03:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e6e2d949fb946e5a33dc03d4ec9f868\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/27054893609e44e58997797cdb9155be : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db3d5946b6114d8887e9e458de81f1c0 : FAILURE in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df45f2066ed74118aa2049615d079ed7 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2490ea8e9ef4a9aa313b6d7c96398f4 : FAILURE in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9cdb7026ef24479a6752b29f0325026 : FAILURE in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff4f813e906d4b69b54274d42d8bb111 : FAILURE in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fbe1615d66045cda990957e9010546f : SUCCESS in 10m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/badf05c214f144fdb917772e66781844 : SUCCESS in 5m 38s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/759ae665d6c34abf9f7bdb2f3292c817 : FAILURE in 50m 17s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/34195a0c93344393bf2907b5e0d54610 : FAILURE in 38m 40s","accounts_in_message":[],"_revision_number":6},{"id":"682e2cee1f167f54f314400ccc6d094f8310c683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-08-19 12:11:12.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ca59eb819183dc5d65da527a834ceb8f6e738608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 13:05:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e542ed14b6054420ad158ddfbc5f22bd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/33967d6b2cee43b39a9073833b2f19c3 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae2c1314cb0849eea0a1457a195be491 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9abf06d194a45ad9edd9a420282bf6e : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f32f0c1745584d008b3b8829c7455cb5 : SUCCESS in 6m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/248deaa8a3b84d3792a9251224b1e194 : SUCCESS in 6m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55e15b5038464d3d8a7e319fb3b9588e : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ce3c133b24429e8bfc67ec7f117e88 : SUCCESS in 4m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ddc333832744d3abc1346807ce0d0bb : SUCCESS in 5m 29s\n- masakari-functional-devstack-multinode https://zuul.opendev.org/t/openstack/build/26faa7836d374e9bbb464a130c373705 : FAILURE in 49m 54s\n- masakari-functional-devstack-ipv6 https://zuul.opendev.org/t/openstack/build/a4f6ccf00c294ca78840f2e84223b15a : FAILURE in 38m 13s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a063ea60473a489eefa3842700645451323e7611","revisions":{"8547cb1b45b43527cb71bef6ffeca7c378ede862":{"kind":"REWORK","_number":1,"created":"2025-06-03 07:12:57.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/13/951613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/13/951613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/13/951613/1"}}},"commit":{"parents":[{"commit":"9561dc73b5eb2723d8b1f40abb6f40569ad0a4d7","subject":"Merge \"HA notification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/9561dc73b5eb2723d8b1f40abb6f40569ad0a4d7"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-03 07:09:20.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-03 07:09:30.000000000","tz":480},"subject":"evaluator rules for ha workflow","message":"evaluator rules for ha workflow\n\nIn practice, ram or disk failure etc. could cause one host frequent\nfauliue, and core service( switch or storage etc.) could cause\nlarge-scale failure. In such case, if it triggers high availability\nrecovery workflow, terrible things could happed, and instances info\nor state would be incorrect.\n\nThis patch adds several evaluator rules, so that when frequent\nsingle-point failures and large-scale failure occurs, it could\nautomatically stop the recovery workflow.\n\nBy default,\nif the number of all hosts notifications in one segment exceeds\n3 times in the last 60 minutes, recovery abort.\nif the number of one host notifications exceeds 1 times in the\nlast 60 minutes, recovery abort.\n...\n\nChange-Id: I30f3df8dc889b64679884944084579815f732314\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/8547cb1b45b43527cb71bef6ffeca7c378ede862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/8547cb1b45b43527cb71bef6ffeca7c378ede862"}]},"branch":"refs/heads/master"},"e8e325bbf87802ef77e430ee8d5bd256d3da1968":{"kind":"REWORK","_number":2,"created":"2025-06-11 02:28:24.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/13/951613/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/13/951613/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/13/951613/2"}}},"commit":{"parents":[{"commit":"9561dc73b5eb2723d8b1f40abb6f40569ad0a4d7","subject":"Merge \"HA notification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/9561dc73b5eb2723d8b1f40abb6f40569ad0a4d7"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-03 07:09:20.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-11 02:27:22.000000000","tz":480},"subject":"evaluator rules for ha workflow","message":"evaluator rules for ha workflow\n\nIn practice, ram or disk failure etc. could cause one host frequent\nfauliue, and core service( switch or storage etc.) could cause\nlarge-scale failure. In such case, if it triggers high availability\nrecovery workflow, terrible things could happed, and instances info\nor state would be incorrect.\n\nThis patch adds several evaluator rules, so that when frequent\nsingle-point failures and large-scale failure occurs, it could\nautomatically stop the recovery workflow.\n\nBy default,\n* If the number of all hosts notifications in one segment exceeds\n3 times in the last 1 hour, recovery abort.\nIf the number of one host notifications exceeds 1 time in the\nlast 24 hours, recovery abort.\n...\n\nChange-Id: I30f3df8dc889b64679884944084579815f732314\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/e8e325bbf87802ef77e430ee8d5bd256d3da1968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/e8e325bbf87802ef77e430ee8d5bd256d3da1968"}]},"branch":"refs/heads/master"},"2cf1a7973d74bb83db2dbd3e68568d2bfb921b34":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-06-27 08:01:05.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/13/951613/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/13/951613/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/13/951613/3"}}},"commit":{"parents":[{"commit":"d75d2165890d498cef489a8ce726be868ef6e37c","subject":"Move test base into tests directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/d75d2165890d498cef489a8ce726be868ef6e37c"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-03 07:09:20.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-27 08:00:36.000000000","tz":480},"subject":"evaluator rules for ha workflow","message":"evaluator rules for ha workflow\n\nIn practice, ram or disk failure etc. could cause one host frequent\nfauliue, and core service( switch or storage etc.) could cause\nlarge-scale failure. In such case, if it triggers high availability\nrecovery workflow, terrible things could happed, and instances info\nor state would be incorrect.\n\nThis patch adds several evaluator rules, so that when frequent\nsingle-point failures and large-scale failure occurs, it could\nautomatically stop the recovery workflow.\n\nBy default,\n* If the number of all hosts notifications in one segment exceeds\n3 times in the last 1 hour, recovery abort.\nIf the number of one host notifications exceeds 1 time in the\nlast 24 hours, recovery abort.\n...\n\nBlueprint: evaluator-rules-for-ha-workflow\nChange-Id: I30f3df8dc889b64679884944084579815f732314\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/2cf1a7973d74bb83db2dbd3e68568d2bfb921b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/2cf1a7973d74bb83db2dbd3e68568d2bfb921b34"}]},"branch":"refs/heads/master"},"8786b32e3bfb7124491de4b25eae09f937636984":{"kind":"REWORK","_number":4,"created":"2025-06-27 09:30:32.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/13/951613/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/13/951613/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/13/951613/4"}}},"commit":{"parents":[{"commit":"d75d2165890d498cef489a8ce726be868ef6e37c","subject":"Move test base into tests directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/d75d2165890d498cef489a8ce726be868ef6e37c"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-03 07:09:20.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-27 09:30:03.000000000","tz":480},"subject":"evaluator rules for ha workflow","message":"evaluator rules for ha workflow\n\nIn practice, ram or disk failure etc. could cause one host frequent\nfauliue, and core service( switch or storage etc.) could cause\nlarge-scale failure. In such case, if it triggers high availability\nrecovery workflow, terrible things could happed, and instances info\nor state would be incorrect.\n\nThis patch adds several evaluator rules, so that when frequent\nsingle-point failures and large-scale failure occurs, it could\nautomatically stop the recovery workflow.\n\nBy default,\n* If the number of all hosts notifications in one segment exceeds\n3 times in the last 1 hour, recovery abort.\nIf the number of one host notifications exceeds 1 time in the\nlast 24 hours, recovery abort.\n...\n\nBlueprint: evaluator-rules-for-ha-workflow\nChange-Id: I30f3df8dc889b64679884944084579815f732314\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/8786b32e3bfb7124491de4b25eae09f937636984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/8786b32e3bfb7124491de4b25eae09f937636984"}]},"branch":"refs/heads/master"},"cc07e694d854a077afdb9e3b61132340e704561b":{"kind":"REWORK","_number":5,"created":"2025-08-19 09:09:49.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/13/951613/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/13/951613/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/13/951613/5"}}},"commit":{"parents":[{"commit":"1c699abd60342247f0e84ba3fc56b64a8bd1ffc7","subject":"Merge \"Add note about requirement lower bounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/1c699abd60342247f0e84ba3fc56b64a8bd1ffc7"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-03 07:09:20.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-08-19 09:08:45.000000000","tz":480},"subject":"evaluator rules for ha workflow","message":"evaluator rules for ha workflow\n\nIn practice, ram or disk failure etc. could cause one host frequent\nfauliue, and core service( switch or storage etc.) could cause\nlarge-scale failure. In such case, if it triggers high availability\nrecovery workflow, terrible things could happed, and instances info\nor state would be incorrect.\n\nThis patch adds several evaluator rules, so that when frequent\nsingle-point failures and large-scale failure occurs, it could\nautomatically stop the recovery workflow.\n\nBy default,\n* If the number of all hosts notifications in one segment exceeds\n3 times in the last 1 hour, recovery abort.\nIf the number of one host notifications exceeds 1 time in the\nlast 24 hours, recovery abort.\n...\n\nBlueprint: evaluator-rules-for-ha-workflow\nChange-Id: I30f3df8dc889b64679884944084579815f732314\nSigned-off-by: suzhengwei \u003csuzhengwei@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/cc07e694d854a077afdb9e3b61132340e704561b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/cc07e694d854a077afdb9e3b61132340e704561b"}]},"branch":"refs/heads/master"},"d823617243ee91f20059b639665664cb0c708609":{"kind":"REWORK","_number":6,"created":"2025-08-19 11:12:22.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/13/951613/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/13/951613/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/13/951613/6"}}},"commit":{"parents":[{"commit":"1c699abd60342247f0e84ba3fc56b64a8bd1ffc7","subject":"Merge \"Add note about requirement lower bounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/1c699abd60342247f0e84ba3fc56b64a8bd1ffc7"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-03 07:09:20.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-08-19 11:11:50.000000000","tz":480},"subject":"evaluator rules for ha workflow","message":"evaluator rules for ha workflow\n\nIn practice, ram or disk failure etc. could cause one host frequent\nfauliue, and core service( switch or storage etc.) could cause\nlarge-scale failure. In such case, if it triggers high availability\nrecovery workflow, terrible things could happed, and instances info\nor state would be incorrect.\n\nThis patch adds several evaluator rules, so that when frequent\nsingle-point failures and large-scale failure occurs, it could\nautomatically stop the recovery workflow.\n\nBy default,\n* If the number of all hosts notifications in one segment exceeds\n3 times in the last 1 hour, recovery abort.\nIf the number of one host notifications exceeds 1 time in the\nlast 24 hours, recovery abort.\n...\n\nBlueprint: evaluator-rules-for-ha-workflow\nChange-Id: I30f3df8dc889b64679884944084579815f732314\nSigned-off-by: suzhengwei \u003csuzhengwei@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/d823617243ee91f20059b639665664cb0c708609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/d823617243ee91f20059b639665664cb0c708609"}]},"branch":"refs/heads/master"},"a063ea60473a489eefa3842700645451323e7611":{"kind":"REWORK","_number":7,"created":"2025-08-19 12:11:12.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/13/951613/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/masakari","ref":"refs/changes/13/951613/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/masakari refs/changes/13/951613/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/masakari refs/changes/13/951613/7"}}},"commit":{"parents":[{"commit":"1c699abd60342247f0e84ba3fc56b64a8bd1ffc7","subject":"Merge \"Add note about requirement lower bounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/1c699abd60342247f0e84ba3fc56b64a8bd1ffc7"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-06-03 07:09:20.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-08-19 12:10:49.000000000","tz":480},"subject":"evaluator rules for ha workflow","message":"evaluator rules for ha workflow\n\nIn practice, ram or disk failure etc. could cause one host frequent\nfauliue, and core service( switch or storage etc.) could cause\nlarge-scale failure. In such case, if it triggers high availability\nrecovery workflow, terrible things could happed, and instances info\nor state would be incorrect.\n\nThis patch adds several evaluator rules, so that when frequent\nsingle-point failures and large-scale failure occurs, it could\nautomatically stop the recovery workflow.\n\nBy default,\n* If the number of all hosts notifications in one segment exceeds\n3 times in the last 1 hour, recovery abort.\nIf the number of one host notifications exceeds 1 time in the\nlast 24 hours, recovery abort.\n...\n\nBlueprint: evaluator-rules-for-ha-workflow\nChange-Id: I30f3df8dc889b64679884944084579815f732314\nSigned-off-by: suzhengwei \u003csuzhengwei@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/a063ea60473a489eefa3842700645451323e7611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/masakari/commit/a063ea60473a489eefa3842700645451323e7611"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
