)]}'
{"id":"openstack%2Fmetalsmith~653800","triplet_id":"openstack%2Fmetalsmith~master~I628b22b0951e1ed10acf961cfc4815b70e61de13","project":"openstack/metalsmith","branch":"master","topic":"allocation","hashtags":[],"change_id":"I628b22b0951e1ed10acf961cfc4815b70e61de13","subject":"[WIP] Switch to allocation API","status":"ABANDONED","created":"2019-04-18 15:54:11.000000000","updated":"2019-06-03 07:48:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b8c0dad43730f68ee05f1cf461ed7c52e09309df","_number":653800,"virtual_id_number":653800,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":-1,"date":"2019-04-24 14:15:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2019-04-29 20:17:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 14:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 20:17:39.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"517f2ff353fe37493fe351e01ddfdd8c541c1db0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-04-18 15:54:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f48432cf127514273cddbb63489d4e7ee0abd0c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 16:34:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/00/653800/1/check/requirements-check/45cfe0b/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/00/653800/1/check/openstack-tox-pep8/82e92d8/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/00/653800/1/check/openstack-tox-py27/b7aaed8/ : FAILURE in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/00/653800/1/check/openstack-tox-py36/222d5f0/ : FAILURE in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/00/653800/1/check/openstack-tox-py37/cae7ae0/ : FAILURE in 4m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/653800/1/check/openstack-tox-lower-constraints/306059f/ : FAILURE in 4m 30s\n- openstack-tox-cover http://logs.openstack.org/00/653800/1/check/openstack-tox-cover/1128bb1/ : FAILURE in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/00/653800/1/check/openstack-tox-docs/dd0f69c/html/ : SUCCESS in 4m 15s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/00/653800/1/check/metalsmith-integration-glance-localboot-centos7/13415ef/ : FAILURE in 35m 20s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/00/653800/1/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/abdc3bb/ : FAILURE in 34m 11s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/00/653800/1/check/metalsmith-integration-glance-netboot-cirros-direct/afac552/ : FAILURE in 32m 12s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/00/653800/1/check/metalsmith-integration-http-netboot-cirros-direct-py3/f5ac979/ : FAILURE in 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3b8bca9d5827049b7972b5b1ca72396c2ebe3ac7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-04-23 14:40:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c78fd1eaa488ab9e5c1f0b18807bf1ee11622583","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 18:11:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/00/653800/2/check/requirements-check/d4d1d16/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/00/653800/2/check/openstack-tox-pep8/a3db76f/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/00/653800/2/check/openstack-tox-py27/7bb153f/ : FAILURE in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/00/653800/2/check/openstack-tox-py36/c4149fc/ : FAILURE in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/00/653800/2/check/openstack-tox-py37/286cca9/ : FAILURE in 7m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/653800/2/check/openstack-tox-lower-constraints/385298d/ : FAILURE in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/00/653800/2/check/openstack-tox-cover/3f370d4/ : FAILURE in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/00/653800/2/check/openstack-tox-docs/d2896b4/html/ : SUCCESS in 3m 47s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/00/653800/2/check/metalsmith-integration-glance-localboot-centos7/3604a0e/ : RETRY_LIMIT in 31m 22s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/00/653800/2/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/a3f110c/ : FAILURE in 32m 38s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/00/653800/2/check/metalsmith-integration-glance-netboot-cirros-direct/fd0619f/ : FAILURE in 31m 35s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/00/653800/2/check/metalsmith-integration-http-netboot-cirros-direct-py3/bb496e6/ : POST_FAILURE in 42m 48s","accounts_in_message":[],"_revision_number":2},{"id":"35085a8b91422144626e6f1fa42e73af5e0cac8c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-04-24 13:13:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b5819bdf36d7b64afb78389cb21da569014c85f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:15:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/00/653800/3/check/requirements-check/b61a68c/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/00/653800/3/check/openstack-tox-pep8/6994c55/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/00/653800/3/check/openstack-tox-py27/53c1b14/ : FAILURE in 5m 30s\n- openstack-tox-py36 http://logs.openstack.org/00/653800/3/check/openstack-tox-py36/b75419e/ : FAILURE in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/00/653800/3/check/openstack-tox-py37/7e7bd28/ : FAILURE in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/653800/3/check/openstack-tox-lower-constraints/1221c9f/ : FAILURE in 3m 59s\n- openstack-tox-cover http://logs.openstack.org/00/653800/3/check/openstack-tox-cover/5d6a975/ : FAILURE in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/00/653800/3/check/openstack-tox-docs/d72aab2/html/ : SUCCESS in 4m 04s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/00/653800/3/check/metalsmith-integration-glance-localboot-centos7/8cd1820/ : FAILURE in 52m 20s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/00/653800/3/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/3b5b6ff/ : FAILURE in 32m 36s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/00/653800/3/check/metalsmith-integration-glance-netboot-cirros-direct/d2d66d3/ : FAILURE in 28m 09s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/00/653800/3/check/metalsmith-integration-http-netboot-cirros-direct-py3/635fb28/ : FAILURE in 33m 00s","accounts_in_message":[],"_revision_number":3},{"id":"28c25fffc312ffdfd592ee2237b621401ba09cfc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-03 07:48:18.000000000","message":"Abandoned\n\nReplaced by other patches","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"17ca9dcfa0454d388b019057f283ea4d6c177c64","revisions":{"72f849cdc484844623e52cccbf15b3ffcb5b6eba":{"kind":"REWORK","_number":1,"created":"2019-04-18 15:54:11.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/00/653800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/00/653800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/1"}}},"commit":{"parents":[{"commit":"f69b81320ba6a277b03d06dfbf271738c6a1f5a4","subject":"Repare the CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f69b81320ba6a277b03d06dfbf271738c6a1f5a4"}]}],"author":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-04-18 15:53:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-04-18 15:54:01.000000000","tz":120},"subject":"[WIP] Switch to allocation API","message":"[WIP] Switch to allocation API\n\nChange-Id: I628b22b0951e1ed10acf961cfc4815b70e61de13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/72f849cdc484844623e52cccbf15b3ffcb5b6eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/72f849cdc484844623e52cccbf15b3ffcb5b6eba"}]},"branch":"refs/heads/master"},"4bf4e85c60c21416c5cc9f1267fe62cd9b821295":{"kind":"REWORK","_number":2,"created":"2019-04-23 14:40:56.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/00/653800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/00/653800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/2"}}},"commit":{"parents":[{"commit":"f69b81320ba6a277b03d06dfbf271738c6a1f5a4","subject":"Repare the CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f69b81320ba6a277b03d06dfbf271738c6a1f5a4"}]}],"author":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-04-18 15:53:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-04-23 14:40:46.000000000","tz":120},"subject":"[WIP] Switch to allocation API","message":"[WIP] Switch to allocation API\n\nChange-Id: I628b22b0951e1ed10acf961cfc4815b70e61de13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/4bf4e85c60c21416c5cc9f1267fe62cd9b821295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/4bf4e85c60c21416c5cc9f1267fe62cd9b821295"}]},"branch":"refs/heads/master"},"17ca9dcfa0454d388b019057f283ea4d6c177c64":{"kind":"REWORK","_number":3,"created":"2019-04-24 13:13:39.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/00/653800/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/00/653800/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/00/653800/3"}}},"commit":{"parents":[{"commit":"f69b81320ba6a277b03d06dfbf271738c6a1f5a4","subject":"Repare the CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f69b81320ba6a277b03d06dfbf271738c6a1f5a4"}]}],"author":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-04-18 15:53:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-04-24 13:13:31.000000000","tz":120},"subject":"[WIP] Switch to allocation API","message":"[WIP] Switch to allocation API\n\nChange-Id: I628b22b0951e1ed10acf961cfc4815b70e61de13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/17ca9dcfa0454d388b019057f283ea4d6c177c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/17ca9dcfa0454d388b019057f283ea4d6c177c64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
