)]}'
{"id":"openstack%2Fmetalsmith~662691","triplet_id":"openstack%2Fmetalsmith~master~I25cdae7d17604140f728fdbcfea4110cbd222679","project":"openstack/metalsmith","branch":"master","topic":"allocation","hashtags":[],"change_id":"I25cdae7d17604140f728fdbcfea4110cbd222679","subject":"Use allocation name for hostname instead of a custom field","status":"MERGED","created":"2019-06-03 07:20:18.000000000","updated":"2019-06-11 09:50:21.000000000","submitted":"2019-06-11 09:50:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662691-1560246621988-e5d355c6","meta_rev_id":"737261ab2f0b608cf90ae58ae4628d33e8c72525","_number":662691,"virtual_id_number":662691,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-11 09:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 08:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 08:39:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 09:50:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbbc8706d0ab16eff8824810ba804eaddf56b1cb","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-03 07:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11b5e397a4d2a465a47bff52fdfa8e77a441546d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 08:06:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/91/662691/1/check/requirements-check/1413994/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/1/check/openstack-tox-pep8/52c5a5e/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/1/check/openstack-tox-py27/d4f7fd5/ : FAILURE in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/1/check/openstack-tox-py36/4b90dda/ : FAILURE in 11m 18s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/1/check/openstack-tox-py37/e5499db/ : FAILURE in 7m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/1/check/openstack-tox-lower-constraints/51a32fb/ : FAILURE in 3m 52s\n- openstack-tox-cover http://logs.openstack.org/91/662691/1/check/openstack-tox-cover/03517d4/ : FAILURE in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/91/662691/1/check/openstack-tox-docs/907e6f4/html/ : SUCCESS in 3m 46s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/1/check/metalsmith-integration-glance-localboot-centos7/e3ee98a/ : FAILURE in 39m 43s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/1/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/f80ccee/ : FAILURE in 32m 10s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/1/check/metalsmith-integration-glance-netboot-cirros-direct/a68b7fd/ : FAILURE in 30m 30s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/1/check/metalsmith-integration-http-netboot-cirros-direct-py3/6b186a7/ : FAILURE in 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"cf0b244718eacbc726454fd23838b40034353bac","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-03 08:12:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"849c87b0e24467d686ae76c84bb70847c2f2bb0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 09:54:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/91/662691/2/check/requirements-check/0c115e7/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/2/check/openstack-tox-pep8/49c05fe/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/2/check/openstack-tox-py27/a7cebad/ : FAILURE in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/2/check/openstack-tox-py36/8985b3e/ : FAILURE in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/2/check/openstack-tox-py37/c29d8cf/ : FAILURE in 4m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/2/check/openstack-tox-lower-constraints/17241c5/ : FAILURE in 3m 47s\n- openstack-tox-cover http://logs.openstack.org/91/662691/2/check/openstack-tox-cover/4dcc4ee/ : FAILURE in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/91/662691/2/check/openstack-tox-docs/24f393f/html/ : SUCCESS in 4m 05s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/2/check/metalsmith-integration-glance-localboot-centos7/a039646/ : SUCCESS in 1h 18m 10s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/2/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/fde880e/ : SUCCESS in 42m 41s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/2/check/metalsmith-integration-glance-netboot-cirros-direct/e98322c/ : SUCCESS in 43m 28s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/2/check/metalsmith-integration-http-netboot-cirros-direct-py3/7e01681/ : SUCCESS in 48m 21s","accounts_in_message":[],"_revision_number":2},{"id":"10c33556ba9589ddd68ad721473752347d4ecb43","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-05 10:56:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"43b00ce9e48959a5b4194a70f32fcf542e68fbf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 12:18:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/91/662691/3/check/requirements-check/41bd6c3/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/3/check/openstack-tox-pep8/62ebd0c/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/3/check/openstack-tox-py27/7e6d87c/ : FAILURE in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/3/check/openstack-tox-py36/6d073bf/ : FAILURE in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/3/check/openstack-tox-py37/37c114d/ : FAILURE in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/3/check/openstack-tox-lower-constraints/cc4e3f0/ : FAILURE in 4m 43s\n- openstack-tox-cover http://logs.openstack.org/91/662691/3/check/openstack-tox-cover/56c72b2/ : FAILURE in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/91/662691/3/check/openstack-tox-docs/71cc3fe/html/ : SUCCESS in 3m 28s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/3/check/metalsmith-integration-glance-localboot-centos7/c740f57/ : SUCCESS in 1h 16m 30s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/3/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/9268a3c/ : SUCCESS in 44m 46s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/3/check/metalsmith-integration-glance-netboot-cirros-direct/c48013d/ : SUCCESS in 46m 11s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/3/check/metalsmith-integration-http-netboot-cirros-direct-py3/5aede52/ : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":3},{"id":"762faceea730bdfc110e9ce5c0d50473073acfb8","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-05 12:55:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d030097b5e5f0bc55567540f5af403b6ba71572e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 14:23:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/91/662691/4/check/requirements-check/941e65f/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/4/check/openstack-tox-pep8/f3ae93a/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/4/check/openstack-tox-py27/a8fb249/ : FAILURE in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/4/check/openstack-tox-py36/9f4fdce/ : FAILURE in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/4/check/openstack-tox-py37/d5428ea/ : FAILURE in 5m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/4/check/openstack-tox-lower-constraints/4f77ff8/ : FAILURE in 4m 40s\n- openstack-tox-cover http://logs.openstack.org/91/662691/4/check/openstack-tox-cover/56bb299/ : FAILURE in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/91/662691/4/check/openstack-tox-docs/2e04b4c/html/ : SUCCESS in 5m 45s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/4/check/metalsmith-integration-glance-localboot-centos7/fc8dabf/ : SUCCESS in 1h 23m 27s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/4/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/b1fbe80/ : SUCCESS in 45m 44s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/4/check/metalsmith-integration-glance-netboot-cirros-direct/258ede4/ : SUCCESS in 56m 50s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/4/check/metalsmith-integration-http-netboot-cirros-direct-py3/2869e16/ : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":4},{"id":"8d40497d85281b21d0169af84f3de15436ce517d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-07 09:29:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cbd7f65d51eebd446b7d4df3f4cb107e4353a876","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 10:33:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/91/662691/5/check/requirements-check/6a488d1/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/5/check/openstack-tox-pep8/4de41e7/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/5/check/openstack-tox-py27/61d2c57/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/5/check/openstack-tox-py36/a15c825/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/5/check/openstack-tox-py37/0bac027/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/5/check/openstack-tox-lower-constraints/f4ee8a4/ : SUCCESS in 5m 01s\n- openstack-tox-cover http://logs.openstack.org/91/662691/5/check/openstack-tox-cover/6952825/cover/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/91/662691/5/check/openstack-tox-docs/407ad62/html/ : SUCCESS in 3m 23s\n- build-openstack-releasenotes http://logs.openstack.org/91/662691/5/check/build-openstack-releasenotes/c1fe37b/html/ : SUCCESS in 3m 59s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-localboot-centos7/184798d/ : FAILURE in 1h 01m 05s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/17d7d49/ : SUCCESS in 48m 42s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-netboot-cirros-direct/cf29e44/ : SUCCESS in 44m 21s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-http-netboot-cirros-direct-py3/1af92b0/ : SUCCESS in 47m 02s","accounts_in_message":[],"_revision_number":5},{"id":"f34a79aac0657e2dc656f1d9f240e91f0949a6dc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-07 10:37:20.000000000","message":"Patch Set 5:\n\nrecheck\n\ngeneric timeout?","accounts_in_message":[],"_revision_number":5},{"id":"7ae22548972680c802f54f8eef51fe99b7562ec5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:54:21.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/91/662691/5/check/requirements-check/e33bd5b/ : SUCCESS in 7m 58s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/5/check/openstack-tox-pep8/f16ad60/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/5/check/openstack-tox-py27/6db5cc7/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/5/check/openstack-tox-py36/903c2dd/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/5/check/openstack-tox-py37/b5e761c/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/5/check/openstack-tox-lower-constraints/ad2d52a/ : SUCCESS in 4m 19s\n- openstack-tox-cover http://logs.openstack.org/91/662691/5/check/openstack-tox-cover/0fca661/cover/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/91/662691/5/check/openstack-tox-docs/8deb988/html/ : SUCCESS in 3m 47s\n- build-openstack-releasenotes http://logs.openstack.org/91/662691/5/check/build-openstack-releasenotes/50bf99f/html/ : SUCCESS in 3m 39s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-localboot-centos7/df272aa/ : FAILURE in 1h 11m 35s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/1991e70/ : SUCCESS in 46m 39s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-netboot-cirros-direct/d1bee10/ : SUCCESS in 47m 07s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-http-netboot-cirros-direct-py3/78c9b8c/ : SUCCESS in 46m 17s","accounts_in_message":[],"_revision_number":5},{"id":"dc45566271cca4c8fc171b4498cb69002e94ea9b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-08 10:59:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"23fb264a4a9131cc76e5f9d76676ff90c40a4561","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 12:46:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/91/662691/5/check/requirements-check/4346110/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/5/check/openstack-tox-pep8/a12dd4f/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/5/check/openstack-tox-py27/3912ec9/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/5/check/openstack-tox-py36/ec6a773/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/5/check/openstack-tox-py37/b27244e/ : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/5/check/openstack-tox-lower-constraints/c150517/ : SUCCESS in 4m 53s\n- openstack-tox-cover http://logs.openstack.org/91/662691/5/check/openstack-tox-cover/c78b8a9/cover/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/91/662691/5/check/openstack-tox-docs/81493f6/html/ : SUCCESS in 4m 22s\n- build-openstack-releasenotes http://logs.openstack.org/91/662691/5/check/build-openstack-releasenotes/e0c24b2/html/ : SUCCESS in 4m 05s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-localboot-centos7/b679fea/ : SUCCESS in 1h 19m 59s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/b78ec54/ : SUCCESS in 44m 54s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-netboot-cirros-direct/33c482d/ : SUCCESS in 48m 44s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-http-netboot-cirros-direct-py3/cf70fe0/ : FAILURE in 47m 26s","accounts_in_message":[],"_revision_number":5},{"id":"b5ca1d24ed9b8b4a07ddfe8a2c99368d91d9c62d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-09 14:17:07.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c9d03fbba5dea9912f7ecc12d089b36063ddcea9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 15:35:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/91/662691/5/check/requirements-check/0c44ebe/ : SUCCESS in 3m 55s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/5/check/openstack-tox-pep8/1e9c102/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/5/check/openstack-tox-py27/31c1a80/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/5/check/openstack-tox-py36/977cf17/ : SUCCESS in 5m 20s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/5/check/openstack-tox-py37/14e310c/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/5/check/openstack-tox-lower-constraints/3104a9e/ : SUCCESS in 6m 00s\n- openstack-tox-cover http://logs.openstack.org/91/662691/5/check/openstack-tox-cover/1311842/cover/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/91/662691/5/check/openstack-tox-docs/13fabdf/html/ : SUCCESS in 4m 57s\n- build-openstack-releasenotes http://logs.openstack.org/91/662691/5/check/build-openstack-releasenotes/0ed640d/html/ : SUCCESS in 5m 08s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-localboot-centos7/2ea2638/ : SUCCESS in 1h 15m 01s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-netboot-cirros-iscsi-py3/0d7b1d5/ : SUCCESS in 43m 52s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/5/check/metalsmith-integration-glance-netboot-cirros-direct/9333a8a/ : SUCCESS in 45m 03s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/5/check/metalsmith-integration-http-netboot-cirros-direct-py3/e5a7107/ : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":5},{"id":"428801f4965391d905240d1d28a02e8004784e54","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-11 08:39:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"57759c5274b1fe3eaced539265e20aaa67ae9738","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 08:39:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"411724d984724c467871caae5493bcb153c09372","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 09:50:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/91/662691/5/gate/requirements-check/ad78f70/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/91/662691/5/gate/openstack-tox-pep8/494e413/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/91/662691/5/gate/openstack-tox-py27/79603a6/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/91/662691/5/gate/openstack-tox-py36/b2e4fc8/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/91/662691/5/gate/openstack-tox-py37/a824bf9/ : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/662691/5/gate/openstack-tox-lower-constraints/64c0349/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/91/662691/5/gate/openstack-tox-docs/ce6b188/html/ : SUCCESS in 4m 28s\n- build-openstack-releasenotes http://logs.openstack.org/91/662691/5/gate/build-openstack-releasenotes/719d294/html/ : SUCCESS in 4m 14s\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/91/662691/5/gate/metalsmith-integration-glance-localboot-centos7/2737668/ : SUCCESS in 1h 08m 06s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 http://logs.openstack.org/91/662691/5/gate/metalsmith-integration-glance-netboot-cirros-iscsi-py3/be42f29/ : SUCCESS in 39m 55s\n- metalsmith-integration-glance-netboot-cirros-direct http://logs.openstack.org/91/662691/5/gate/metalsmith-integration-glance-netboot-cirros-direct/4b6d356/ : SUCCESS in 46m 49s\n- metalsmith-integration-http-netboot-cirros-direct-py3 http://logs.openstack.org/91/662691/5/gate/metalsmith-integration-http-netboot-cirros-direct-py3/841a433/ : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":5},{"id":"b680cc6c6846c4121abe69897ba8a16d011f1bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 09:50:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e9c25b02e538440481792d2ec7eaa1303c639ede","revisions":{"819881150cd46dd2d1c7a0d8714ed043919dd77b":{"kind":"REWORK","_number":1,"created":"2019-06-03 07:20:18.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/91/662691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/91/662691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/1"}}},"commit":{"parents":[{"commit":"b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a","subject":"Merge \"Split InstanceConfig into GenericConfig and CloudInitConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-03 06:39:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-03 07:19:56.000000000","tz":120},"subject":"[WIP] Use allocation name for hostname instead of a custom field","message":"[WIP] Use allocation name for hostname instead of a custom field\n\nChange-Id: I25cdae7d17604140f728fdbcfea4110cbd222679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/819881150cd46dd2d1c7a0d8714ed043919dd77b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/819881150cd46dd2d1c7a0d8714ed043919dd77b"}]},"branch":"refs/heads/master"},"0f0e1e7de082309e240c0b0cb560b6cfac821114":{"kind":"REWORK","_number":2,"created":"2019-06-03 08:12:56.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/91/662691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/91/662691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/2"}}},"commit":{"parents":[{"commit":"b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a","subject":"Merge \"Split InstanceConfig into GenericConfig and CloudInitConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-03 06:39:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-03 08:12:48.000000000","tz":120},"subject":"[WIP] Use allocation name for hostname instead of a custom field","message":"[WIP] Use allocation name for hostname instead of a custom field\n\nChange-Id: I25cdae7d17604140f728fdbcfea4110cbd222679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0f0e1e7de082309e240c0b0cb560b6cfac821114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0f0e1e7de082309e240c0b0cb560b6cfac821114"}]},"branch":"refs/heads/master"},"7d0e48c5ac836a204f617bbea129b03385766002":{"kind":"REWORK","_number":3,"created":"2019-06-05 10:56:15.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/91/662691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/91/662691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/3"}}},"commit":{"parents":[{"commit":"b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a","subject":"Merge \"Split InstanceConfig into GenericConfig and CloudInitConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-03 06:39:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-05 10:56:03.000000000","tz":120},"subject":"[WIP] Use allocation name for hostname instead of a custom field","message":"[WIP] Use allocation name for hostname instead of a custom field\n\nChange-Id: I25cdae7d17604140f728fdbcfea4110cbd222679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7d0e48c5ac836a204f617bbea129b03385766002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7d0e48c5ac836a204f617bbea129b03385766002"}]},"branch":"refs/heads/master"},"884b019e6d26c711afaa3a956851b2830cbde049":{"kind":"REWORK","_number":4,"created":"2019-06-05 12:55:08.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/91/662691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/91/662691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/4"}}},"commit":{"parents":[{"commit":"b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a","subject":"Merge \"Split InstanceConfig into GenericConfig and CloudInitConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-03 06:39:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-05 12:54:54.000000000","tz":120},"subject":"[WIP] Use allocation name for hostname instead of a custom field","message":"[WIP] Use allocation name for hostname instead of a custom field\n\nChange-Id: I25cdae7d17604140f728fdbcfea4110cbd222679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/884b019e6d26c711afaa3a956851b2830cbde049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/884b019e6d26c711afaa3a956851b2830cbde049"}]},"branch":"refs/heads/master"},"e9c25b02e538440481792d2ec7eaa1303c639ede":{"kind":"REWORK","_number":5,"created":"2019-06-07 09:29:20.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/91/662691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/91/662691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/91/662691/5"}}},"commit":{"parents":[{"commit":"b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a","subject":"Merge \"Split InstanceConfig into GenericConfig and CloudInitConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b1769b01cc7deeb837cfbdf2dbcf08fa90ba669a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-03 06:39:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-07 09:08:48.000000000","tz":120},"subject":"Use allocation name for hostname instead of a custom field","message":"Use allocation name for hostname instead of a custom field\n\nThis concludes the switch to the allocation API.\n\nChange-Id: I25cdae7d17604140f728fdbcfea4110cbd222679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e9c25b02e538440481792d2ec7eaa1303c639ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e9c25b02e538440481792d2ec7eaa1303c639ede"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
