)]}'
{"id":"openstack%2Fmetalsmith~685100","triplet_id":"openstack%2Fmetalsmith~master~Ia2f29fd924095b1a9c0cbd988878901dcb313b85","project":"openstack/metalsmith","branch":"master","topic":"gate","hashtags":[],"change_id":"Ia2f29fd924095b1a9c0cbd988878901dcb313b85","subject":"Lower MTU","status":"ABANDONED","created":"2019-09-26 14:32:57.000000000","updated":"2019-10-18 10:51:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19ed8012d03afab915bc32c59e5250e0ca894e10","_number":685100,"virtual_id_number":685100,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-26 16:15:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-09-27 11:10:20.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 16:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ec2a1ab31ba8b3d0a16f345936a77c05cdb4059","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-26 14:32:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ed716b7bf10f78e0929b6be9dd327a41d2cc218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:15:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efed0ded0ef848c0aae00659f1190556 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81e9ac13a94d491b8f7471b1b43b8148 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb0ef26418bd495187bea4dec3e51b98 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ca2a686bfb8414090275b5ded9ba6a3 : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5b6e0ccaed94cafa034c1ea887b158d : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/839670f283f6458eae3fc41f1c6797af : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00087fb22f334db5bbe33f298c8ff994 : SUCCESS in 6m 01s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/290559a37dd7428599ac3f19b0e2b040 : FAILURE in 1h 19m 21s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/b0fcb03a51e54570b2086f7a7d52f22f : SUCCESS in 48m 17s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/6c88cfbee3b14842b187f00bc5c777cf : SUCCESS in 58m 03s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/966238333f934012b4521907fc7062e9 : SUCCESS in 48m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d81cd0e8d9dc21bf4ae9e8ffabbf4519a8bd7320","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-27 11:10:20.000000000","message":"Patch Set 1:\n\nrecheck\n\ndoesn\u0027t look good..","accounts_in_message":[],"_revision_number":1},{"id":"3323905d4c544a639fe2c34a7837cb646a903206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 12:33:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38419a28aa164f61aea4a855525f8fa4 : SUCCESS in 2m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/feb6b576dfa442f4abd4d31f6999b2f2 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c12372afa614913aeff70622da8d219 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acc6ad33c5144307bf6b072661d6ed09 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8732a7028f4b439da19a2fc587efc08d : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fe94c4efbaa4d998f0f908ba40636ed : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8bbb775ea343fb998d8889fbf43857 : SUCCESS in 4m 40s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/36bb2c14135544eaae686341192c78e7 : FAILURE in 1h 07m 43s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/37309270d6624f4d85beb7fd4d3fedee : SUCCESS in 46m 31s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/a24a0f7b37a34196988c0895c4744bd5 : SUCCESS in 46m 26s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/97c97520ec2247c9951fb123ef7f7bd0 : SUCCESS in 54m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e4392a9e13e321a82d3fcbfa95ec54368b2c730f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-10-18 10:51:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9b46164c3fe76d5a12d3397949f1f1c2572078d","revisions":{"a9b46164c3fe76d5a12d3397949f1f1c2572078d":{"kind":"REWORK","_number":1,"created":"2019-09-26 14:32:57.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/00/685100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/00/685100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/685100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/685100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/685100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/00/685100/1"}}},"commit":{"parents":[{"commit":"2d9b7cadf5faa9d6343a013b2989a3d83e6d75aa","subject":"Merge \"Build pdf doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2d9b7cadf5faa9d6343a013b2989a3d83e6d75aa"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-26 14:32:20.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-26 14:32:20.000000000","tz":120},"subject":"Lower MTU","message":"Lower MTU\n\nFor some unknown reason, new networks being created in the\nfortnebula test VMs are defaulting to a MTU size of 1350 bytes.\n\nThis is based upon the physnet MTU, which does appear to be\nlower on the fortnebula cloud.\n\nChange-Id: Ia2f29fd924095b1a9c0cbd988878901dcb313b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a9b46164c3fe76d5a12d3397949f1f1c2572078d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a9b46164c3fe76d5a12d3397949f1f1c2572078d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
