)]}'
{"id":"openstack%2Fmetalsmith~685911","triplet_id":"openstack%2Fmetalsmith~master~I786f52f02b12c4543b5f6c48a4f4d6e8f7df900a","project":"openstack/metalsmith","branch":"master","topic":"test","hashtags":[],"change_id":"I786f52f02b12c4543b5f6c48a4f4d6e8f7df900a","subject":"DNM testing the CI","status":"ABANDONED","created":"2019-10-01 09:13:26.000000000","updated":"2019-10-18 10:51:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41c090e8fc908ab8e7c8ad1e5bf90e157ebc4c27","_number":685911,"virtual_id_number":685911,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-09 10:14:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-09 14:55:07.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 10:14:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b045c287dfd17192cc91b291aae6ea8bb6863ad8","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-10-01 09:13:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf5e37b59748065ac039cdad11e1c104121c0310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 10:39:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/138452143a3c4f36b18bb5565a11b4e0 : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8cf874393824d81b4430c7ea6313edf : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33de009cbdb24170b07713a682c75050 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ba6974b7a084ec3b21a260e08df5a50 : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/017a4f1d0c5348c4bcbfd593ddd7b5d4 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16084ece417744848f117a75457ac04f : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d052d47aca2f48648bee19552977ba3b : SUCCESS in 5m 02s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/6df6e453af364008bb8ef77e50a0d1b1 : FAILURE in 1h 17m 45s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/2c622cb61dba4e53b73ef4784611bead : FAILURE in 1h 12m 31s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/cfae7be4aae3430395d38d5650a0cd81 : SUCCESS in 48m 43s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/8a16b989a057466580da7d4c63d7f206 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e0602e61fc31bb6d27f45be03434caf6c3721125","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-10-08 15:21:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffb6db36316151e6668867adac6027206571500e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 17:00:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624b9e9005b64bf4b69f845418a56a01 : SUCCESS in 2m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/079b0dea0bbd4d2ea10ec9a68b17ba35 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c56cca1c9c8b4dbbaabf22bfa5a8ec4e : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15faa2aac313494cb90514ded36f6827 : SUCCESS in 3m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0020050e696742a8a4c270c7c63de5c3 : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec9d0d860af540519f8c2580f714e53f : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d69bd3e913141d99c7cbd9426604807 : SUCCESS in 4m 53s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/724de29aa8d84a5688b4c723a4053721 : FAILURE in 1h 22m 52s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/9d0cd1a6fde044bf917b3345c55e1292 : SUCCESS in 56m 38s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/d9b6ec0f76fd4c16850e26a785a29c23 : SUCCESS in 1h 10m 32s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/aa9d0221cd9f447b854cee32d80df69a : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":2},{"id":"24df5e0d7dd1f341454a80c1ca92012b232ee110","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-10-09 08:45:20.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"03148bc7d6745113eae2356f5cf24e271c5bca6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 10:14:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f162886875b45e2bace206136d3cc48 : SUCCESS in 2m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e9cdc33f0fa4b27a8cf7635bb5c83f7 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04298458e0a545e78c7d04ef67798216 : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e04d594fbae469d81b42d13c931e60b : SUCCESS in 2m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c333e27be29467887c1f607a717b3a2 : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcb47da8e43c401e975db9403f3ba02d : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d42d779182448ce87e1e7ad0e1f97ca : SUCCESS in 4m 36s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/f4cf798664bb4f5cbac29241ffcdad66 : FAILURE in 1h 23m 28s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/7bf436b02e064c4c984e992a2db70c19 : SUCCESS in 40m 20s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/a0c3c8623ae542928f27403a2b1d4b22 : SUCCESS in 39m 56s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/b2400ab24e1d42e786c6808f109ada80 : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":3},{"id":"c708c6a28c9ce4b72f4393fbaa598f3d428597c1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-10-09 14:55:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5835d51ae59f0bc6742a13446de49c8d7db66dcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:24:11.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e64c8f1d1b2e4856aae6138b1f0e60a1 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e9bbf3d021944f0858b0115bde21b56 : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2113bc6ef1e54cd29341c3b401243e51 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc4a6317c9924c63b798c192b98d1884 : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b998c9bc5d24ac888ade8598131452b : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eca3078114a342b38ea11720cd12713a : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a6495b75f1492e87e070ff5056e18c : SUCCESS in 11m 26s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/1b1121b7bdcf4f83a0d2b63a69534ae6 : FAILURE in 1h 25m 33s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/d82ce099b5c346d1ad876002205f3441 : SUCCESS in 58m 07s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/0e23e37c4a2444b5afcab4de3a216dd4 : SUCCESS in 52m 49s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/0ed04b1fd2ff41778ed81e9143a3f967 : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":3},{"id":"46a53976f92cc3fc6c3ba30e52aafe79da64709b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-10-18 10:51:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe16c4515483b9a84f4909746fc8b2f88677e09c","revisions":{"7aec6c99ac3bcc24eceee63f23090a2e28b736b3":{"kind":"REWORK","_number":1,"created":"2019-10-01 09:13:26.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/11/685911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/11/685911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/1"}}},"commit":{"parents":[{"commit":"2d9b7cadf5faa9d6343a013b2989a3d83e6d75aa","subject":"Merge \"Build pdf doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2d9b7cadf5faa9d6343a013b2989a3d83e6d75aa"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-10-01 09:12:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-10-01 09:12:54.000000000","tz":120},"subject":"DNM testing the CI","message":"DNM testing the CI\n\nDepends-On: https://review.opendev.org/#/c/683798/\nChange-Id: I786f52f02b12c4543b5f6c48a4f4d6e8f7df900a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7aec6c99ac3bcc24eceee63f23090a2e28b736b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7aec6c99ac3bcc24eceee63f23090a2e28b736b3"}]},"branch":"refs/heads/master"},"51eb8edd257975fdfa7c394522373b2b4ff33ef5":{"kind":"REWORK","_number":2,"created":"2019-10-08 15:21:54.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/11/685911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/11/685911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/2"}}},"commit":{"parents":[{"commit":"a9b46164c3fe76d5a12d3397949f1f1c2572078d","subject":"Lower MTU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a9b46164c3fe76d5a12d3397949f1f1c2572078d"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-10-01 09:12:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-10-08 15:21:37.000000000","tz":120},"subject":"DNM testing the CI","message":"DNM testing the CI\n\nDepends-On: https://review.opendev.org/#/c/683798/\nChange-Id: I786f52f02b12c4543b5f6c48a4f4d6e8f7df900a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/51eb8edd257975fdfa7c394522373b2b4ff33ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/51eb8edd257975fdfa7c394522373b2b4ff33ef5"}]},"branch":"refs/heads/master"},"fe16c4515483b9a84f4909746fc8b2f88677e09c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-09 08:45:20.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/11/685911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/11/685911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/11/685911/3"}}},"commit":{"parents":[{"commit":"a9b46164c3fe76d5a12d3397949f1f1c2572078d","subject":"Lower MTU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a9b46164c3fe76d5a12d3397949f1f1c2572078d"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-10-01 09:12:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-10-09 08:44:48.000000000","tz":120},"subject":"DNM testing the CI","message":"DNM testing the CI\n\nDepends-On: https://review.opendev.org/#/c/685324/\nChange-Id: I786f52f02b12c4543b5f6c48a4f4d6e8f7df900a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/fe16c4515483b9a84f4909746fc8b2f88677e09c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/fe16c4515483b9a84f4909746fc8b2f88677e09c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
