)]}'
{"id":"openstack%2Fmetalsmith~689108","triplet_id":"openstack%2Fmetalsmith~master~Ie8dadff9fc5df180effd6016e464be280bbf64fa","project":"openstack/metalsmith","branch":"master","topic":"test","hashtags":[],"change_id":"Ie8dadff9fc5df180effd6016e464be280bbf64fa","subject":"DNM testing the CI","status":"ABANDONED","created":"2019-10-17 09:31:27.000000000","updated":"2019-10-18 10:51:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"79614433a015e2c63d345c43ae3d09df00f4eb3d","_number":689108,"virtual_id_number":689108,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-17 11:10:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 11:10:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad99b1a4bc5cc71ee609bc994e619d95a1e6d85c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-10-17 09:31:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59f77a62bf266240f022227c5d14c7e710ec3eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 11:10:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408177cd12d0457f962b95aa1a7fd9b2 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7011ff82ef3444c199cb4d7382882f99 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bc4055571de48aeaa0361d11eb511e7 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10f0b3221f834968aeeffc0d64ac91da : SUCCESS in 3m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b7ca8e725149bf994c4399d73dc85e : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfcfd05a5b024b3486c391f19bdb58c7 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab7b2d18dfe473a9a678260baa032c3 : SUCCESS in 5m 54s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/64df7389a8014d13953ce01be1364fba : RETRY_LIMIT in 24m 48s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/5a60813c20d7472cb436e6a44e4d903e : RETRY_LIMIT in 27m 14s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/95a4f1a790f1473680859e9d64d44cf5 : RETRY_LIMIT in 26m 42s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/5f1696e7973d4c52adbc2489f5d4583f : RETRY_LIMIT in 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bc1c4a7f2bb9bc4e4a9f5b8a8331538ae6d4caa9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-10-18 10:51:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdb6bf1a0d45d0b363bd496fa2828bcd02a8bb5b","revisions":{"bdb6bf1a0d45d0b363bd496fa2828bcd02a8bb5b":{"kind":"REWORK","_number":1,"created":"2019-10-17 09:31:27.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/08/689108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/08/689108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/08/689108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/08/689108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/08/689108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/08/689108/1"}}},"commit":{"parents":[{"commit":"3c95b4ac4277b3ee64cd44d5f00e94dec365c1ea","subject":"Add versions to release notes series","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/3c95b4ac4277b3ee64cd44d5f00e94dec365c1ea"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-10-01 09:12:54.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-10-17 09:31:12.000000000","tz":120},"subject":"DNM testing the CI","message":"DNM testing the CI\n\nChange-Id: Ie8dadff9fc5df180effd6016e464be280bbf64fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/bdb6bf1a0d45d0b363bd496fa2828bcd02a8bb5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/bdb6bf1a0d45d0b363bd496fa2828bcd02a8bb5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
