)]}'
{"id":"openstack%2Fmetalsmith~693999","triplet_id":"openstack%2Fmetalsmith~stable%2Ftrain~I4460755ee66b4aa0b8a651b6bd142c769d669ee2","project":"openstack/metalsmith","branch":"stable/train","topic":"root-size-stable/train","hashtags":[],"change_id":"I4460755ee66b4aa0b8a651b6bd142c769d669ee2","subject":"Stop requiring root size for whole disk images","status":"MERGED","created":"2019-11-13 07:59:13.000000000","updated":"2020-01-20 10:27:44.000000000","submitted":"2020-01-20 10:25:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693999-1579515945809-9981773a","meta_rev_id":"62ac9e14cf25f9d646752c1c215a8372921c2eb5","_number":693999,"virtual_id_number":693999,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 10:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-01-20 08:59:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"date":"2020-01-17 12:03:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2020-01-20 08:59:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 12:03:27.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-01-20 08:59:34.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-01-20 10:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d405182a55add6ea3a7a5a7ddd8f3af93e23dd93","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-13 07:59:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8bb0e69dae1810c132d5b1c2f5182d1845f25701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 09:28:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aad1b1fd575493c9d8cf4b79be0880f : SUCCESS in 2m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/896096a26b284e6f8b3bc0ed7fd46d34 : SUCCESS in 2m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a11d7f796fc24073983ba945c3cf23d6 : SUCCESS in 2m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e202f0c74b64852866639400d04a022 : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96723d5bd40c40fca04c4ae52c20314e : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecb9e684969941caa4fc7c2778250305 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c8f4d681e248f981b0117fd1845dfe : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c96329e137fd4835bb701b890a133cfb : SUCCESS in 2m 53s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/4d6d6261d3604668ad2e35bbb846b819 : SUCCESS in 1h 23m 06s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/e70cfa489ad54583a3aa3c118b78bda8 : SUCCESS in 47m 29s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/24bcc447ec1742ff97ab6168050d4ba2 : SUCCESS in 49m 53s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/d38bd4386fd64f5a9e1873652915f531 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ae74a8ebe5a8075ffb61d132e98df3078715f002","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-20 08:59:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4df304b5dad21e9ffa14fdacae954c08a1bc27d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 08:59:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c17ed8ad7ae2a611f4e2f979f00297f3e49a0b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 10:25:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de726866ddc4dd2ab053f949968dab4 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c157e1e9141248b79b0af96ec5a358eb : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40f3bd75b24c4211a70fcaa59b61504d : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78c1674bd6e7495ab270a35d0d9527c1 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40325eadc02942c5ac848448c60246af : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aface03783304c9395814e915f9cf340 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0198cb265e01486e82b64085c0174113 : SUCCESS in 5m 49s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/cb317397e011441c9d52a3748b484053 : SUCCESS in 1h 20m 53s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/8d30622237634cb8b42e5be56f4ce557 : SUCCESS in 48m 53s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/cfae1a991fb84f59aad61e145e5c038d : SUCCESS in 49m 17s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/f49c0445b3ab4ded8de142650b20740c : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"70ac5cd93c0a2d1d137ad4d4c7eb840ca0170370","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 10:25:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5990f9891008c2cdabe5d955988d2f875c80854c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 10:27:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/096c6c4808214830ac318a2e237dbce4 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44cc5abb977a4733ba0183e15c1b7ae2 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0afed7462dfd962ce7061b662ace72132b475c4d","revisions":{"0afed7462dfd962ce7061b662ace72132b475c4d":{"kind":"REWORK","_number":1,"created":"2019-11-13 07:59:13.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/99/693999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/99/693999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/99/693999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/99/693999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/99/693999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/99/693999/1"}}},"commit":{"parents":[{"commit":"749d403c19afa8517cfa61b42c6d1b24a9cd2402","subject":"Build pdf doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/749d403c19afa8517cfa61b42c6d1b24a9cd2402"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-26 11:57:18.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-11-13 07:59:13.000000000","tz":0},"subject":"Stop requiring root size for whole disk images","message":"Stop requiring root size for whole disk images\n\nThis requirement has been fixed in newer versions of ironic.\n\nChange-Id: I4460755ee66b4aa0b8a651b6bd142c769d669ee2\n(cherry picked from commit 2d801e2526f4d1dd9968a4a3e4ab6a8603e1416d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0afed7462dfd962ce7061b662ace72132b475c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0afed7462dfd962ce7061b662ace72132b475c4d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
