)]}'
{"id":"openstack%2Fmetalsmith~695727","triplet_id":"openstack%2Fmetalsmith~master~Icf19488a3f230084faaf2ab5ece626b03495592e","project":"openstack/metalsmith","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Icf19488a3f230084faaf2ab5ece626b03495592e","subject":"Drop python 2.7 support and testing","status":"MERGED","created":"2019-11-22 17:12:04.000000000","updated":"2019-11-25 18:17:53.000000000","submitted":"2019-11-25 18:14:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695727-1574705666713-7789a045","meta_rev_id":"6966ff9049b1560c361afddab5ce105742b85bb6","_number":695727,"virtual_id_number":695727,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 18:14:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-11-25 16:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-25 16:43:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 16:38:42.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-11-25 16:43:50.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-11-25 18:14:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4a13546ff4072dfff8c1cad045b60847566046b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-11-22 17:12:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfd50eecf8dcdd873d2ca3f0868dac5d111c795e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 19:09:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36546f9ac36d45ee8fa92aa501800703 : SUCCESS in 2m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/704bfde053fc44fe90e7eddb744f10d4 : FAILURE in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9794eed4294726b6c2bd76e4615f09 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82564956d74045598609670a7e925eb1 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aae958fb8bf8417897a81fa300e2a492 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b599ce072094e2a80bca2e38a0ced1a : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e29b40667cb4b58a1ebf034a98deabd : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8972813d831741848d8f73c2e811f4b5 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8124704fbf464ed5aa5346596bc70202 : SUCCESS in 3m 07s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/157ef19b39324f50afd3ecf12ae023a9 : SUCCESS in 1h 51m 02s\n- metalsmith-integration-glance-netboot-cirros-iscsi-py3 https://zuul.opendev.org/t/openstack/build/17b4710391e24c14845667e4294d7dc1 : SUCCESS in 52m 04s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/41e83f1ce6a045b1b5dc2c4027c75c37 : SUCCESS in 58m 19s\n- metalsmith-integration-http-netboot-cirros-direct-py3 https://zuul.opendev.org/t/openstack/build/d780d0f6d4044e06b89b2fd9583d38c3 : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5e0b6faf0a22bb01564a11fbe1e0e7419318d9bb","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-22 22:57:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fded8021bbfcdfc01d873428b4feb3aba8f81fdf","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-11-25 10:06:30.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"348e46db0d95824cdba3c44420fdb15750dff9b1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-25 10:10:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"541f8dd65b7a1a33248912aa89abc9e7f7b3d725","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-11-25 10:20:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbc029f5d4626d068cf6bd524a385bea7f13ebc5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-25 11:21:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a337ee2104c03f472508bf4339ae5f137e7e7f4f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-25 11:35:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c674b318a828390479e5b05c5c0822aabfe2869d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 11:44:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0609828f1e7144bdacff5a92ff714f36 : SUCCESS in 1m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/57cc8626967d4d1e905a15d70553f826 : FAILURE in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b234d07190cb4467b421d35dc4a44a86 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f20b4f05ade24b4bbfdb9e0c2fa07723 : SUCCESS in 2m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c787088850724caab9dbafd7b355ae26 : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78feb6515cef44a3842d927ec1e7c4c6 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b844c3e757543a5977f487623f9f576 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3210f25f57044a1cafde5f25246a77d0 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/444de0b0f06b4eefbb330da95ab097a3 : SUCCESS in 2m 43s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/4945110b80bb406d851334f7ebb9d269 : SUCCESS in 1h 18m 28s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/356c5dca10d04c65af11581022ff3206 : RETRY_LIMIT in 5m 44s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/cecae315db814a998202ab00662d3e7e : SUCCESS in 48m 38s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/57feff841e8f4c2bb54a6183b26d4802 : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":2},{"id":"deedd58b66825f94e28c12827d31b44f681fa966","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-11-25 12:33:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"062ef79e555d31931fd9166368598c9ab49c34b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:00:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd61e9f017b9415299e129a579f4739c : SUCCESS in 1m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c0013a98ff44d9a9d03f9bff880b496 : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25f79a4c9ba4d92b45d0389441f2670 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/373080b5ed6a471599b61539693f8996 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24419dd00f24409390a039554cd48172 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cd43009231242e98affb1e39c463f02 : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b656ec1b83643c783ba812603c2535a : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a908306b226a4cb7b7b1273c472ba218 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b282ec758ebf407596d0f6d73e76982b : SUCCESS in 4m 46s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/60d299aa91f34a7a88c969af2e49adf8 : SUCCESS in 1h 22m 35s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/87346b3e3651449395b065b633b77390 : SUCCESS in 46m 20s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/b7b7a6664bc44c38b8bf6e554a1a3a2b : SUCCESS in 51m 36s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/4b4b5dc1e05b44c8b0b032060ba74667 : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":3},{"id":"1a55363da50553db0cb64c6254d2c44de67f5818","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-25 16:38:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9fc8014cb0cc44a51db98e0fa521d8b92600aae6","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-25 16:43:50.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6cc42133282b225d397148d0a6749f0e8a08706f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:44:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9ecef2bc37b1ccabce120175fb533ae39ad903a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:14:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/410c80ca63444c25a2d9c09ca2f8202e : SUCCESS in 2m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/33fe0924880043cf8b50cc175f3da58e : SUCCESS in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35d1e8a6ea14a219cb93720c9a15de7 : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a9f4ebcc4d4bdc9bb2c5ed8aed7b61 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8615227936fb478c87d13d46718afccb : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b53cf7cbfc24e9aae7c0a07a86b6934 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309285c81be2460f9620390ac66c3870 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86088799d03e49af90016ed6cbe7366f : SUCCESS in 3m 47s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/b706a8bf6e5b4b87ad19973970e2d59d : SUCCESS in 1h 26m 46s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/3a939ffc4aee4ace9e8405b794352ec0 : SUCCESS in 54m 38s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/3faaf111914842b390fc46968a4322f5 : SUCCESS in 1h 07m 34s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/21810fa9cd07462081e311dccf3e8efe : SUCCESS in 1h 06m 52s","accounts_in_message":[],"_revision_number":3},{"id":"30e2ba9c5f6ec2d8ad58401eb76026807e070481","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:14:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ce5d44b8dab915dc78f4ce9bcdd3b3c9169351cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:17:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fa7cb8cb8449a78efff40834fc3d78 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd08aa7da97145229ba33940ae17c859 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"33871b187b60ff904f4560c89310b44b53ea7548","revisions":{"f1fb2e9d99de2a422fb5104444932aa2a329736d":{"kind":"REWORK","_number":1,"created":"2019-11-22 17:12:04.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/27/695727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/27/695727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/1"}}},"commit":{"parents":[{"commit":"2d801e2526f4d1dd9968a4a3e4ab6a8603e1416d","subject":"Stop requiring root size for whole disk images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2d801e2526f4d1dd9968a4a3e4ab6a8603e1416d"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-11-22 17:10:54.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-11-22 17:10:54.000000000","tz":60},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nmetalsmith is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Icf19488a3f230084faaf2ab5ece626b03495592e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f1fb2e9d99de2a422fb5104444932aa2a329736d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f1fb2e9d99de2a422fb5104444932aa2a329736d"}]},"branch":"refs/heads/master"},"00792bb34ca885429e42adda98d17b38c7a36efc":{"kind":"REWORK","_number":2,"created":"2019-11-25 10:20:34.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/27/695727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/27/695727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/2"}}},"commit":{"parents":[{"commit":"2d801e2526f4d1dd9968a4a3e4ab6a8603e1416d","subject":"Stop requiring root size for whole disk images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2d801e2526f4d1dd9968a4a3e4ab6a8603e1416d"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-11-22 17:10:54.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-11-25 10:20:25.000000000","tz":60},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nmetalsmith is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Icf19488a3f230084faaf2ab5ece626b03495592e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/00792bb34ca885429e42adda98d17b38c7a36efc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/00792bb34ca885429e42adda98d17b38c7a36efc"}]},"branch":"refs/heads/master"},"33871b187b60ff904f4560c89310b44b53ea7548":{"kind":"REWORK","_number":3,"created":"2019-11-25 12:33:46.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/27/695727/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/27/695727/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/27/695727/3"}}},"commit":{"parents":[{"commit":"2d801e2526f4d1dd9968a4a3e4ab6a8603e1416d","subject":"Stop requiring root size for whole disk images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2d801e2526f4d1dd9968a4a3e4ab6a8603e1416d"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-11-22 17:10:54.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-11-25 12:33:37.000000000","tz":60},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nmetalsmith is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Icf19488a3f230084faaf2ab5ece626b03495592e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/33871b187b60ff904f4560c89310b44b53ea7548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/33871b187b60ff904f4560c89310b44b53ea7548"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
