)]}'
{"id":"openstack%2Fmetalsmith~737886","triplet_id":"openstack%2Fmetalsmith~master~I34114e653a0714cb5991672af1192e9fad95ea6b","project":"openstack/metalsmith","branch":"master","topic":"metalsmith_instances","hashtags":[],"change_id":"I34114e653a0714cb5991672af1192e9fad95ea6b","subject":"Move roles into a standard directory layout","status":"MERGED","created":"2020-06-24 23:13:54.000000000","updated":"2020-08-16 17:54:51.000000000","submitted":"2020-08-16 17:53:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737886-1597600429022-39e11680","meta_rev_id":"3736add4eeb0b870e227f37008b792aaf4878e60","_number":737886,"virtual_id_number":737886,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-16 17:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-08-10 18:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-16 16:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-16 16:28:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 18:59:22.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-08-16 16:28:52.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-16 17:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d384d65a1fcb9fe8d51d2422cdc9e9c9ff2d7b5","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-24 23:13:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bfe1c1d31b8caa3867983cdf2fbebe516dc5c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:03:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9cf2e93a96f14dd7a3b1889560c5a4d6 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c326983f327d4d01b462ac334a317222 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2a680b41cab4ee0af210311b679421e : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f30cc4e2b6c0411891310a217ba44bd4 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fb096a1a27945888c5d2098ab0cac57 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd359a83e7664cf987a71fdaa27130b7 : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ff7cd6b960b45c2949efd1b5cc1163a : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea4911b74a94d5dba173a77142f72b6 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb07546fe7f84243bbf7b9e92e193f05 : SUCCESS in 4m 21s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/dfd70337b9a34a8cbd074ca6242b6ee9 : FAILURE in 40m 56s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/ae73d851d1fc4286a43a569567963f2b : FAILURE in 32m 04s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/3e16d1c22a644dd9bc3e655c66359d4f : FAILURE in 33m 54s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/569aaf47cbd84570a06d907fedce622a : FAILURE in 31m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a9458794a38b1b8c0fcc546bdf0ac23661bf2e1f","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-25 03:28:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de74a3ccd479af60506abec2a2371a50405f33f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 04:55:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9ad23d4922c14d9fa92740a12f7fd745 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6142d43efe064e55952d1b802b7ab0e3 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/737da19499794c5d980aeb806c857f01 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78a4021bd17a44699cd780ad324607b9 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fc401f022b7492a84495a1552d90304 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0977710b20d84a4095742c6baaa823fd : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c964e04a35054c4f88775b9d4e3066d7 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0446787bb0c84b18b9f93d42fca06001 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3dbaaa7094e4013a78fad2d962fd870 : SUCCESS in 4m 09s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/423e2a0b7f434013945d05daa23c1cd6 : SUCCESS in 1h 21m 29s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/2c2c05cc012944458291a6955cebfc31 : SUCCESS in 53m 04s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/44c139aa93d24e239cfe70d6cdf0cd06 : SUCCESS in 54m 35s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/2be5e03d86f84671a20fc33cf2c42240 : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":2},{"id":"c196fc30b9b4932a2b677fed1d06e05aa0a479ba","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-25 07:48:53.000000000","message":"Patch Set 2: Code-Review-1\n\nLet\u0027s leave CI playbooks alone, locating them in ./playbooks is a zuul convention (and we shouldn\u0027t distribute them).","accounts_in_message":[],"_revision_number":2},{"id":"56299860243d68892a68dc5a4c7ca9f0512eec4e","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-25 23:09:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"119a03335c41718790efb68caca46c0d5f1d1d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 00:53:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/25cab541b19a431592e5b80203173ce8 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d857c15d4b474f9741b8a3046ab2b0 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2869f56acd8d4ca9a9dda72e000fc19e : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9037aa11b8e74fafb207383731f05d6a : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d25122d813346abafaaff3a50e3fdae : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f2b94c174ea4a97aaeb3af98296c1db : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0636c943a8942a1ab6c125198c1bcc3 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b66b1b7bdf6c45f59b6355684a4431c9 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af850cb38c304758933c70f7c8e4e73a : SUCCESS in 4m 22s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/3b96200591ac40e5922b47c910d401b3 : SUCCESS in 1h 36m 58s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/02c87373f6b64827a2375660903e3683 : SUCCESS in 48m 33s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/b5eba286efc14108807186a5c9c39145 : SUCCESS in 57m 48s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/acdab459c9794f6097d736a1d98a5c8c : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c243d7d0f971e41c42e7774a95141a323514aa61","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-07-01 01:38:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"db6b55f15c58edc555b310cffb3ff57a246cc7ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 03:08:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce5ede34513f4ec69e310690d3ce1d37 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a424198119451a87d1f44e7918f8f5 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68feada6b26c4176be6217e0eb559025 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/984b5a5aae354e5999ead249ac86eee3 : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7f8829cb9b74ed28d3a98759a54816d : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adf66203865e44a9bb1758c9a43f72d0 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68fe104ac33a4e88a8a070b55c17bfdf : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16775d084727414190493b60691d7215 : SUCCESS in 4m 18s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c97df53425e940d78065aa80deab3104 : SUCCESS in 1h 28m 30s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/11560c1b1c9a4cc28054ed20630d2cb2 : SUCCESS in 57m 02s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/da0400b0688c4ee685d0910b507b2fe6 : SUCCESS in 1h 04m 33s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/8c33d3150627438f894b804754abcfb6 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":4},{"id":"cb6878271e2e51b3706dee76bef393a0f1072aba","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-10 18:59:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9e9ace5c8052fba362e6a983aeb79ca190d31371","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-16 16:28:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"413918c6c31ea59003b8e7b61b28cf1bd7fddee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 16:31:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"55e82c6dd1dd1ff96d14bc129b1acd2491caa9ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 17:53:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/810510a972104aec86cfe03e98ad9530 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a0eeb0ffc0435b87c030b64b89c1d1 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20706f1dec3e40c997730dd91b8f3286 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3bdb951f62945ee99eda56aa8316ac1 : SUCCESS in 3m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a03ad7f406554606b632fe1dbdd18ac2 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb9f65faca14aa48212d6c9d46841f0 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/014cb429ecaf452485806ea37b011449 : SUCCESS in 3m 02s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/740dda75af964dd284b440d76c7fc5a7 : SUCCESS in 1h 20m 41s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/34d615994ba8432ea7508caf7641cebc : SUCCESS in 56m 11s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/29304cdf11174d95a3c9bc5bfad3cdae : SUCCESS in 37m 45s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/f1c6c5c0c1e440cda042377294758e65 : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":4},{"id":"66c7a9ca36f9c1b68d4164d9cfe649d533d432f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 17:53:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"61ebbd3140018cf299e1e502f18cb3aa203fdcbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 17:54:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e74aa4292c9433d8bd69f6b9e44d967 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbc41fbb7e944d5194161c30a9a434c5 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"53636ba03c89f08fe00f2b0fa24b9cba2aa5cd10","revisions":{"baf8b0a1b86efc038b5edf39894bc8cd8bcfffbd":{"kind":"REWORK","_number":1,"created":"2020-06-24 23:13:54.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/86/737886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/737886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/1"}}},"commit":{"parents":[{"commit":"35e0457b6a4845fef5a887f280c5fc5d133469ea","subject":"Use is instead of ambiguous pipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/35e0457b6a4845fef5a887f280c5fc5d133469ea"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:15:01.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 03:59:27.000000000","tz":720},"subject":"Move roles and playbooks into a standard directory layout","message":"Move roles and playbooks into a standard directory layout\n\nThis change adopts the same directory layout as tripleo-ansible[1]\nin preparation for the metalsmith repo containing ansible plugins such\nas modules.\n\n[1] https://opendev.org/openstack/tripleo-ansible/src/branch/master/tripleo_ansible\n\nChange-Id: I34114e653a0714cb5991672af1192e9fad95ea6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/baf8b0a1b86efc038b5edf39894bc8cd8bcfffbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/baf8b0a1b86efc038b5edf39894bc8cd8bcfffbd"}]},"branch":"refs/heads/master"},"ac01f0d2691f459d9ffb04add42526dc0369b9a7":{"kind":"REWORK","_number":2,"created":"2020-06-25 03:28:42.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/86/737886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/737886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/2"}}},"commit":{"parents":[{"commit":"35e0457b6a4845fef5a887f280c5fc5d133469ea","subject":"Use is instead of ambiguous pipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/35e0457b6a4845fef5a887f280c5fc5d133469ea"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:15:01.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-25 03:27:12.000000000","tz":720},"subject":"Move roles and playbooks into a standard directory layout","message":"Move roles and playbooks into a standard directory layout\n\nThis change adopts the same directory layout as tripleo-ansible[1]\nin preparation for the metalsmith repo containing ansible plugins such\nas modules.\n\nLike in tripleo-ansible, the roles directory becomes a symlink to the\nnew roles location so that ansible can discover the role without\nhaving to install.\n\n[1] https://opendev.org/openstack/tripleo-ansible/src/branch/master/tripleo_ansible\n\nChange-Id: I34114e653a0714cb5991672af1192e9fad95ea6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ac01f0d2691f459d9ffb04add42526dc0369b9a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ac01f0d2691f459d9ffb04add42526dc0369b9a7"}]},"branch":"refs/heads/master"},"31011060874f9afe0a3dccaa1fcab8c8a54ed94b":{"kind":"REWORK","_number":3,"created":"2020-06-25 23:09:05.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/86/737886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/737886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/3"}}},"commit":{"parents":[{"commit":"35e0457b6a4845fef5a887f280c5fc5d133469ea","subject":"Use is instead of ambiguous pipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/35e0457b6a4845fef5a887f280c5fc5d133469ea"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:15:01.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-25 22:37:32.000000000","tz":720},"subject":"Move roles into a standard directory layout","message":"Move roles into a standard directory layout\n\nThis change adopts the same directory layout as tripleo-ansible[1]\nin preparation for the metalsmith repo containing ansible plugins such\nas modules.\n\nLike in tripleo-ansible, the roles directory becomes a symlink to the\nnew roles location so that ansible can discover the role without\nhaving to install.\n\n[1] https://opendev.org/openstack/tripleo-ansible/src/branch/master/tripleo_ansible\n\nChange-Id: I34114e653a0714cb5991672af1192e9fad95ea6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/31011060874f9afe0a3dccaa1fcab8c8a54ed94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/31011060874f9afe0a3dccaa1fcab8c8a54ed94b"}]},"branch":"refs/heads/master"},"53636ba03c89f08fe00f2b0fa24b9cba2aa5cd10":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-01 01:38:40.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/86/737886/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/737886/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/737886/4"}}},"commit":{"parents":[{"commit":"72072cd5099e25baf4e88425413545f897624b1e","subject":"Remove ussuri jobs in favor of victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/72072cd5099e25baf4e88425413545f897624b1e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:15:01.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-07-01 01:30:35.000000000","tz":720},"subject":"Move roles into a standard directory layout","message":"Move roles into a standard directory layout\n\nThis change adopts the same directory layout as tripleo-ansible[1]\nin preparation for the metalsmith repo containing ansible plugins such\nas modules.\n\nLike in tripleo-ansible, the roles directory becomes a symlink to the\nnew roles location so that ansible can discover the role without\nhaving to install.\n\n[1] https://opendev.org/openstack/tripleo-ansible/src/branch/master/tripleo_ansible\n\nChange-Id: I34114e653a0714cb5991672af1192e9fad95ea6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/53636ba03c89f08fe00f2b0fa24b9cba2aa5cd10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/53636ba03c89f08fe00f2b0fa24b9cba2aa5cd10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
