)]}'
{"id":"openstack%2Fmetalsmith~737887","triplet_id":"openstack%2Fmetalsmith~master~I02407547154d8d6084fa30c5fe164c5b6a060043","project":"openstack/metalsmith","branch":"master","topic":"metalsmith_instances","hashtags":[],"change_id":"I02407547154d8d6084fa30c5fe164c5b6a060043","subject":"Move metalsmith_instances from tripleo-ansible","status":"MERGED","created":"2020-06-24 23:13:54.000000000","updated":"2020-08-27 22:18:11.000000000","submitted":"2020-08-25 11:38:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737887-1598355530857-1d2acb16","meta_rev_id":"8b9b063499fb01c45ec4c81701bccf235a5de7b8","_number":737887,"virtual_id_number":737887,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 11:38:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-08-10 19:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-25 10:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-25 10:04:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 19:17:42.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-08-25 10:04:46.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-25 11:38:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b5a5fb610e1d4601b1a8706bdcf785139f63a29","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-24 23:13:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7a098b2d63123f8a6c9e7a546ee33807e639515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:05:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e36a1af5afc4b92a878710ef980a6c4 : SUCCESS in 3m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a8b0bbf7de441ab89a1476d652a7675 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c9306743e3742cdbd4a8a48cd23d171 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9c16dc3a8834d6c84be672d655568d8 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2f0973386f44fa6bff36ed47801cfe6 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eeb989b12b84bb4a44d436a60a0dc4b : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44a6230cb3b04c9aad661caa316ddee8 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17c8eb78925047d7aa38f11ab11144a8 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e2987102964e919ad16774b5ae2784 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6439d7a4c7424521bd14906943dc7de0 : SUCCESS in 4m 39s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/eeca2f1036aa45eebdc593dbb9548ea3 : FAILURE in 41m 30s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/889fc3bb058c43dfbc2ef0e141f28573 : FAILURE in 35m 32s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/dcdc2ffd2cde43cc811ecc765a286f5c : FAILURE in 33m 51s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/7b4ed1fa5f89427ea3b3d933efa337de : FAILURE in 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"84e588f11892bf410fac2437f7015175e846e28f","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-25 03:28:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0a53d86441b8958340505a66639d9401dab04b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 04:46:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0441e3cbec524edbbc56eeafaec1ec0f : SUCCESS in 2m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99d652782e6a47309b706b84a61b6c7d : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceab509732c6453da50d65ea8374b03c : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5157765dc804cbd8b27e77b80aba15a : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adb583916bd94d898c06dc758cae2a75 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f908b5c7a5e743459bff88eedb77f82f : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab415a8f56cc4ec1b880ea0c9d0b212f : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/730bf0ff78814831a183797d768d0992 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f5b6390a984b88aef9934efc6b83dc : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c845130e64141b4b89c68c142bc8f2e : SUCCESS in 3m 34s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/6a2d0a9eea9d43aa8799be2fa11bc1ad : SUCCESS in 1h 15m 52s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/ba06e6601f96495393303898fcdc80b2 : SUCCESS in 36m 19s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/334233632db647d195e0d7a00f9938f6 : SUCCESS in 58m 41s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/9d59af73ae3a46b685a6f5dea01bff94 : SUCCESS in 48m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4ce4db396866a51893282930e759d4ced942ef16","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-25 23:09:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"941a8cf0ca106240cd07b89a601aad84b5383504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 00:30:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1a521f796d4549acb913780c784a6313 : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed3f4f676c444dac9cdae97af51b6616 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2298f6c0251341a798e8dd5963f77819 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8890b044564e7e8529bc1c95f72256 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d59f35161d40414d92b0a5ec4c1f3c95 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef7cd6ade3e440179f6cdb4e0569c28e : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4d772f22f0749c2bd362f9c9867ef51 : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea654e2adf784894a0f60901537c8d84 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c55d83aec94d23942f1878f58a3412 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4001b85679624b67bbd709e65e0e5842 : SUCCESS in 3m 39s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/1bc1325c79644be08a5c438d6fc9fa5c : SUCCESS in 1h 17m 12s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/f1eaab6ea803474ab6e61fadde8195ff : SUCCESS in 50m 45s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/42e3dc2fb2474c1a8142439c54f4021f : SUCCESS in 1h 07m 09s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/a693f3b1e44c4a2f97d3e01e167083ba : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":3},{"id":"8d49b99d5ec879e15e6897d923c1696956681c02","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-28 22:42:48.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5d0d4f5281b460045e3673cc99761f7faecbf4c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 00:00:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cebee03f214741948988e408610bc520 : SUCCESS in 1m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e064eb9198a1497ea27ef6a6aa9ef34a : SUCCESS in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c84bec6fbec4a8b8a06ad05cd74b58e : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c2d00774d0b4d46a93c7a9c5290a367 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3831904d256746cca3e30a4877affc4d : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5ce325a7e8448b78d330bf537f8db68 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f2ba8d5667143a4bfe1194ab30fbf21 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658f0edd6f9c4dd8b9e3e529d435e46f : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22f4c22fac4c4d03928e78685b69794e : SUCCESS in 2m 49s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/a1dc519c909444099e53ef7bd6af552e : SUCCESS in 1h 09m 45s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/cd4ce0c980034725acf19b793ec84f46 : SUCCESS in 47m 26s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/f85e6d7455c04883b2077d47e5db2720 : SUCCESS in 52m 15s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/967e920e263140baa6ff1d543fdafb2c : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":4},{"id":"a0d1d4fd5bc93d79392a4112ae541279c25f6c76","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-07-01 01:38:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"987b691acd2e7f40171020183903fda26d724103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 03:14:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/159301c44c8048abaf8c040927652848 : SUCCESS in 3m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/50f3bf16b7a0480cab52e3d50e8309c2 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef2d293dca0e4065b3afa73d4a701255 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d8d92281858440ea916c8a383913fc1 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d4f6d2bbdb24ce28ca724a8986d5d5e : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10bfe98f874141bfaa045f0f512dd5ed : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a092ab6808745b0b61fc06d31c9e649 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08249273b25e4de98439bbfcd91f0007 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09ebb6ec55c14707987614a038c91fe2 : SUCCESS in 4m 44s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/1099f7f8cd274e0ead9efe15db0e8ae0 : SUCCESS in 1h 28m 41s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/42e51dce9a1341518e24b4e5dc926793 : SUCCESS in 1h 09m 42s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/7ce59b42dea642c58180341bb1bb34b3 : SUCCESS in 1h 03m 56s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/9065c56c6ee44da28b24bfca5367c656 : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":5},{"id":"4274965e90df94bf9aaae816afafbcec62e688de","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-10 19:17:42.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4a23ad0dcd5d3b7d9d0f620deb6cfd9969e9ec12","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-16 16:30:31.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"36a2a9e21171b04150c71e2c67c3ac4d0a797005","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-25 10:04:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(5 comments)\n\nApproving for the sake of meeting the release deadlines but please do check our comments inline.","accounts_in_message":[],"_revision_number":5},{"id":"aaa5502f18be1ad20424f400ba01ad86064d0d61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 10:05:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4f63d459dff216f6db4f10a839b7448a292f05d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:38:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0591c2daac1f469b9cfe80336fa9b7fb : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d918f7cfb2da49fbbc77302ab7f32054 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f9b65634c4431f8c199ba361a4be81 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4741b6e805c402791eb3d40aba1131b : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7071363d28144fe7a1bcf47afd68f03d : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3204785ca65471f9ad1be2aa833f4af : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3a2589ebfa47c29461653ec1a2a74e : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7052ceef36f547d998f3955311e09b09 : SUCCESS in 3m 50s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/622ff733f77f48e8a2076d71234e17b9 : SUCCESS in 1h 28m 05s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/049017620b274812b1bb782eaf6b19a6 : SUCCESS in 50m 46s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/4cc756c2bf994417a2473602fa228c68 : SUCCESS in 55m 37s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/dc8718f18093439a8a981c95166bda0f : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":5},{"id":"21f81b440adf65fc17ea9531283741d7691f03e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:38:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"73552309fab941984b8471028af8fb38249d5714","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:40:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6534dda13ef046d9af714a054f015b32 : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9fa87a7e4f1449d97291dd3fc36abe2 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5},{"id":"225622ad8a771193920e8995882668d3881bd122","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-08-27 22:18:11.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"62b8b1d8b9697248d4a3368e921b9ee67b73fd0c","revisions":{"d99aec47965c3710651e1753ed3d0694c52b3895":{"kind":"REWORK","_number":1,"created":"2020-06-24 23:13:54.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/87/737887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/87/737887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/1"}}},"commit":{"parents":[{"commit":"baf8b0a1b86efc038b5edf39894bc8cd8bcfffbd","subject":"Move roles and playbooks into a standard directory layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/baf8b0a1b86efc038b5edf39894bc8cd8bcfffbd"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:40:54.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 23:09:27.000000000","tz":720},"subject":"WIP Move metalsmith_instances from tripleo-ansible","message":"WIP Move metalsmith_instances from tripleo-ansible\n\nMarked as WIP because the reviews for the next change in this series\nmight result in changes to metalsmith_instances, and if the changes\nare large it might be better to do them in tripleo-ansible before\nmoving the module.\n\nChange-Id: I02407547154d8d6084fa30c5fe164c5b6a060043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d99aec47965c3710651e1753ed3d0694c52b3895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d99aec47965c3710651e1753ed3d0694c52b3895"}]},"branch":"refs/heads/master"},"b7e8e5054ce0ed814f7648b954415e521f3199ee":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-25 03:28:42.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/87/737887/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/87/737887/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/2"}}},"commit":{"parents":[{"commit":"ac01f0d2691f459d9ffb04add42526dc0369b9a7","subject":"Move roles and playbooks into a standard directory layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ac01f0d2691f459d9ffb04add42526dc0369b9a7"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:40:54.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-25 03:28:27.000000000","tz":720},"subject":"WIP Move metalsmith_instances from tripleo-ansible","message":"WIP Move metalsmith_instances from tripleo-ansible\n\nMarked as WIP because the reviews for the next change in this series\nmight result in changes to metalsmith_instances, and if the changes\nare large it might be better to do them in tripleo-ansible before\nmoving the module.\n\nChange-Id: I02407547154d8d6084fa30c5fe164c5b6a060043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b7e8e5054ce0ed814f7648b954415e521f3199ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b7e8e5054ce0ed814f7648b954415e521f3199ee"}]},"branch":"refs/heads/master"},"f5bd0cd140763bf5bbd617b220600e96cac5a6c1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-25 23:09:05.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/87/737887/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/87/737887/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/3"}}},"commit":{"parents":[{"commit":"31011060874f9afe0a3dccaa1fcab8c8a54ed94b","subject":"Move roles into a standard directory layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/31011060874f9afe0a3dccaa1fcab8c8a54ed94b"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:40:54.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-25 22:38:00.000000000","tz":720},"subject":"WIP Move metalsmith_instances from tripleo-ansible","message":"WIP Move metalsmith_instances from tripleo-ansible\n\nMarked as WIP because the reviews for the next change in this series\nmight result in changes to metalsmith_instances, and if the changes\nare large it might be better to do them in tripleo-ansible before\nmoving the module.\n\nChange-Id: I02407547154d8d6084fa30c5fe164c5b6a060043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f5bd0cd140763bf5bbd617b220600e96cac5a6c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f5bd0cd140763bf5bbd617b220600e96cac5a6c1"}]},"branch":"refs/heads/master"},"473c801532621a1fc69d9c845853e0d02dfecda8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-28 22:42:48.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/87/737887/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/87/737887/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/4"}}},"commit":{"parents":[{"commit":"31011060874f9afe0a3dccaa1fcab8c8a54ed94b","subject":"Move roles into a standard directory layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/31011060874f9afe0a3dccaa1fcab8c8a54ed94b"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:40:54.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-28 22:42:12.000000000","tz":720},"subject":"Move metalsmith_instances from tripleo-ansible","message":"Move metalsmith_instances from tripleo-ansible\n\nChange-Id: I02407547154d8d6084fa30c5fe164c5b6a060043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/473c801532621a1fc69d9c845853e0d02dfecda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/473c801532621a1fc69d9c845853e0d02dfecda8"}]},"branch":"refs/heads/master"},"62b8b1d8b9697248d4a3368e921b9ee67b73fd0c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-01 01:38:40.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/87/737887/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/87/737887/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/87/737887/5"}}},"commit":{"parents":[{"commit":"53636ba03c89f08fe00f2b0fa24b9cba2aa5cd10","subject":"Move roles into a standard directory layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/53636ba03c89f08fe00f2b0fa24b9cba2aa5cd10"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 01:40:54.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-07-01 01:30:35.000000000","tz":720},"subject":"Move metalsmith_instances from tripleo-ansible","message":"Move metalsmith_instances from tripleo-ansible\n\nChange-Id: I02407547154d8d6084fa30c5fe164c5b6a060043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/62b8b1d8b9697248d4a3368e921b9ee67b73fd0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/62b8b1d8b9697248d4a3368e921b9ee67b73fd0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
