)]}'
{"id":"openstack%2Fmetalsmith~737888","triplet_id":"openstack%2Fmetalsmith~master~Ia30620821182c58050813e807cdde50a27d03c15","project":"openstack/metalsmith","branch":"master","topic":"metalsmith_instances","hashtags":[],"change_id":"Ia30620821182c58050813e807cdde50a27d03c15","subject":"metalsmith_deployment role switch to metalsmith_instances","status":"MERGED","created":"2020-06-24 23:13:54.000000000","updated":"2020-10-08 21:11:33.000000000","submitted":"2020-10-08 21:09:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737888-1602191386918-8f1aa1ae","meta_rev_id":"997dd4fd087cd83b90343be01f50983ada00745d","_number":737888,"virtual_id_number":737888,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-08 21:09:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-07 23:27:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-10-07 20:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2020-10-08 19:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 09:25:39.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-10-07 23:27:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 19:56:38.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-10-08 21:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"115bf0285c95cfb795e62d11c08014561e266ad8","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-24 23:13:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d84e952d9dcae9bef1694ad5c00575137e9a4294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:07:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dcda2b743644d4cb46ef739fab8ad89 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd1c21e21ef645baa919ffa950d49cdb : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/110fa2fa32f64a20afad480df1e5f704 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/474b1d2381054d7fb9d9947c7e13fa4f : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3715b34a064d440d9b1d5bddc49cb2a3 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae1795b8c0a646ee9c92fad850f82877 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d9f745577e4dbb9df6584c34807d67 : SUCCESS in 6m 19s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/12f1e4bd9310456083413223384ebb9a : FAILURE in 42m 25s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/289fcb7a78af422cbfdb22b8e9241a62 : FAILURE in 33m 54s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/fa0dd87cad79482baab353be62fbb156 : FAILURE in 37m 31s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/d90454c5e0354722b536b0bfd8bcc979 : FAILURE in 38m 59s","accounts_in_message":[],"_revision_number":1},{"id":"be897a59ee6b86627a1733828b8f67ca5877ddd4","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-25 03:28:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"65f2d145be8ae0e830970be6b4808c121c5a13eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 04:18:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb13fcde0c684b0c866b46fecbc4a373 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16020b08f7164296833346a6318bed1d : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ed868f0c1a4467d938f5763bea18a94 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4891fd62d85044e596a51837d27d866d : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1b8839e8d6d41d69ad34d267c6e0a7c : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/438f4a90f933405d8630317fbcce9298 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f9ff49cd7a4a7ca16b4e3cc3e4ee03 : SUCCESS in 7m 29s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/87cefaebfcb24d739a8a3eb92b67b45e : FAILURE in 44m 10s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/a6574597f6c043d586ccd8bd04940bfc : FAILURE in 34m 41s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/9434ac9cdf424d6da3ed9fe647b319e4 : FAILURE in 22m 12s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/1e15ef59a76242f9a6d9036c005694ee : FAILURE in 32m 36s","accounts_in_message":[],"_revision_number":2},{"id":"584d74857c5e22549d3282d313ba3129aa355776","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-25 23:09:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"03360cb1751f38e45c0ec50689407a9ca08f7002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 23:53:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30674ee6079641d5aa857dd2c6f65cf2 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6729928f6ad4c3997e1952054263bad : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9e344689fb642bfa595f79b3d077dac : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c109e4ce5c484fea825bc4f6ae7e7485 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2ed443192db48e1b3c0eaf9d534ffd4 : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fd099b9965f43d5b25692ea925cefbb : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa66d53afc7e4abebfbcf6599ffee3b0 : SUCCESS in 8m 02s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/f11bf8bca65a4ad5bf500922946890e1 : FAILURE in 38m 17s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/fdf2092aefc147bdbd86cfe8b640fcc1 : FAILURE in 29m 13s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/8bb8935f5364434498b201140889e637 : FAILURE in 28m 55s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/b16e61bf86744bb59c6fec9dcd6781cb : FAILURE in 34m 27s","accounts_in_message":[],"_revision_number":3},{"id":"5866858674893636cd6bc1318104aa8a7afde877","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-26 01:49:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fb86f1f663655938eb9647cefc4ef96e76eb634e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 02:37:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec83dea92a244b884a9a6e8bf317392 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19279c9ebd24f579b92ab1ba9f40cee : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d34d794fd6994b3ba5b1bd00a3b52ee1 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12728450c0a94f858bf1928b7238766a : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/932880b42e3f4c78933513c1e91f3afa : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7d6623342f1489ab69c90753d1f2004 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43779eff5f394c96b88709ff1ad95c72 : SUCCESS in 6m 38s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/d699fbca79224437890459f00136d650 : FAILURE in 42m 38s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/64c1fdd202204706a082409301fe4e30 : FAILURE in 28m 38s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/ba95071285d04ebba34e1f3aa973010e : FAILURE in 34m 34s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/ecb23173f7a74aa594cbd3828aefc7aa : FAILURE in 21m 43s","accounts_in_message":[],"_revision_number":4},{"id":"76f280274b54923d2dacc0ea563d6e24d8cd0450","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-26 03:18:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f4500108d36dd580b6376d8f974e90925d927da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 04:33:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91aae072c55b4816b3feaf4e57383605 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12676359e00e4a8f9e9a507803a66fca : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4590ec7a449647c28e4be31ddc1a37b7 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15fecae41bfb4187a84d897106a3ed51 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efcb1218a9af44b3ba67f9c29bbcc566 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e5fe5263c74f649e159721f6ac87f5 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ccdd2d01a3403fbe302da5d580459c : SUCCESS in 5m 02s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/b9d9e96d6d99475abf38edda16dedabd : FAILURE in 1h 09m 10s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/9d5c365a1d404d378d97b04adb453db5 : FAILURE in 48m 31s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/936d70f569da46f4844a831266016bb3 : FAILURE in 52m 32s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/9c4d0c26a5db4558a5c0eed2ae2fba3d : FAILURE in 50m 46s","accounts_in_message":[],"_revision_number":5},{"id":"bcbacfc28a8ee4cb5c22244187ed181a97fcdd78","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-28 22:42:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f2ca9513d494e65e4f24736da90a050d5a1326f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 23:47:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68f97abe82354a3d9dc8b48064c2cb1d : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b94b5a9d98c4f7e8bbf9585feed23d7 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44715771fbf14a84845807b7a34d12b1 : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4c9cf636db14f92aa1d8d7254100189 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f2936ebb183455a935092f75ea22020 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29114d7411f41e1acd33963b071580f : SUCCESS in 6m 34s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/fdc322a309fd4999b1b9c0d23ee92cb6 : FAILURE in 58m 45s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/3b890ed7b4db4a0b8690ea86b46cad56 : FAILURE in 45m 44s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/10d31388af6e4cfb9a182a79f75ec3f2 : FAILURE in 46m 03s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/5a4f7fc9c0eb4292bbc9d7242ad44ae8 : FAILURE in 46m 09s","accounts_in_message":[],"_revision_number":6},{"id":"c8d3838537629f94ba1e499700f47d8b9887c2e8","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-29 01:36:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9b5daff5fc88fff372edda72ae88ee5695c823d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 02:47:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0666927037fe46d98a9bcf44a2ba129f : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d35c4e83f874eedbe811ad3a6292c8c : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f50e4590ae7402da0a7e38b8d9e264d : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdd73078ac0f472eb0d93f88d429d0b9 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef0c31b49c8c4f61955193c701b3b301 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e3ffc3b25a748f1b7427fbb158c1861 : SUCCESS in 6m 03s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/5c23bd19243041bf92f567fecef09767 : FAILURE in 1h 10m 26s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/6eface0ab8fe4bde9d9a310ba45af957 : FAILURE in 48m 57s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/5a77daf805904db98dab2d054988e84a : FAILURE in 45m 12s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/4897ea0343a14e4faf33f9b85df7129b : FAILURE in 49m 57s","accounts_in_message":[],"_revision_number":7},{"id":"1b5665d3b0850d1bb1099b8a484e4d57ffabd37f","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-06-29 04:30:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1fc7af42fea46bec4677ec899095b60709cf3487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 05:39:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fed3716768c4ed5a3f2e20919535fcc : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/315a9fcdbe024d219f9609884d8c3a73 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/595b25733da64f3caa1a9141e9484e14 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a2f5c36202d44499b601a2870133cc9 : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a836e135142946b2a782b9a715dfa008 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb006c00d1e4d0f99ce9585726fb14c : SUCCESS in 6m 03s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/a0ad692020d6490a95d1578d6e673449 : FAILURE in 1h 07m 22s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/786427a77b2846f68d7fbd20128fcdbf : FAILURE in 43m 10s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/f4ca67fe7fee4a6f963dc9a57669aa9f : FAILURE in 44m 55s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/c4dcc2c4e8da471cb2f0815ddc3a7fed : FAILURE in 44m 45s","accounts_in_message":[],"_revision_number":8},{"id":"a82d2d8c277adee0b3822214fc8079b0d5552435","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-07-01 01:38:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dcb96ee56fb46ef4168ba2a470d0a8c1358d3022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 02:57:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfff85208ade41c1bd98adc4aac9f2da : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33ee2443819a42f79ed88e8c623cfcfd : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f70b688ca040689c1e61e785665d31 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ebd64040a8444b99223acdd4d240120 : SUCCESS in 5m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d1e4ccec9e242e4aa519e495e64568d : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447bc45f16f94e4c9a8d0a229ffc4831 : SUCCESS in 5m 45s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/ea79f61287bb4da7ad1c3c1596425fa9 : FAILURE in 1h 16m 57s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/06b8bb8077304e439f532686d980aed7 : FAILURE in 1h 00m 04s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/30d4e01e5e334c3995839fc31e0d16d3 : FAILURE in 58m 51s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/7c6022d302694b25b81533ca9c363bc3 : FAILURE in 1h 01m 36s","accounts_in_message":[],"_revision_number":9},{"id":"7b5b8b169fc7a72a704ee6587edefa2dac5fbcbc","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-07-01 22:59:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5dd36adcc93b6b8f77a597ff24f72fd63ee3a24f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 00:10:23.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1050329ce42c41c9b717e146a75d8aa9 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5dbbedf4b10453bb7afb408e0482786 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40dd4716dfb940e5bc6a082a61ef63ae : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3de3eeca2b8f4f12b222ffe1cca84179 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/889ca5a0d4774efd9e9d729625802d07 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1ff00c2faad401186154d7494ed1be7 : SUCCESS in 4m 45s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/23f7f5f6f7a14ce78257e9ff31a74e24 : FAILURE in 1h 06m 06s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/d1a1efcc556549f4992856c50d046f50 : FAILURE in 43m 48s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/6a7b6dfe30aa4aa6af2b50e22dd21171 : FAILURE in 48m 10s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/e0547a6b2f984511922e1027d3dcbd51 : FAILURE in 1h 10m 29s","accounts_in_message":[],"_revision_number":10},{"id":"e963f3775d456ac35ded5cc229bea1ca210b119d","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-07-02 02:17:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"707baa665effbeeb1244f0d11136564f23d24e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 03:30:07.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3feb17f01be34b1e88e8a4a79c950cec : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10f08af5d76a426da40ded87acd52541 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2b72997bbb545d79bcb70cc92bfd99f : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3efdeac1a1c4cef975fa38633c1451e : SUCCESS in 9m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/209b5920f8f64214b5790506c93b932e : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d3c0a1fd02490fb0aa263299971ff7 : SUCCESS in 5m 15s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/b028f9515c4049f6a2452c3e59a74928 : FAILURE in 1h 07m 02s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/fd73a4560b9f408c8a5e07019fee4fbf : FAILURE in 55m 36s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/71aa4414a93e4fafab8f01779d3bf3e7 : FAILURE in 46m 11s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/92262d634c394ab79efa6db9e0509ebb : FAILURE in 57m 51s","accounts_in_message":[],"_revision_number":11},{"id":"21c925ba240f03ad91de0999980dd5ac17a612f5","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-07-02 03:54:41.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"323a84cc33c5744163e8527d1264e283b9c59160","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 04:36:33.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db885d5cba0429b9865284e5d23de01 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbc7c8fbf876473a965b233c7223607a : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18f9dd00a33a4ac590ada719e7a38489 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fa3b55eb75745588107d8f5553a712d : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da79091d5ba147d089623984651702a4 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f93477e3265458cba30756c5d9de72a : SUCCESS in 4m 31s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/f5e46badc6bf4fd0b9361a2e3a5c1393 : FAILURE in 36m 53s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/86bf202ddec340cabfc437aa448c18f0 : FAILURE in 29m 46s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/96780920c20e42728ec12ad67399b03e : FAILURE in 30m 40s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/f8a1fdb13d2748a6bd70c0bb14ba204a : FAILURE in 29m 08s","accounts_in_message":[],"_revision_number":12},{"id":"ce94889aa38cffb7bdb0f393ed84ab8225a95e4a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-25 10:05:33.000000000","message":"Patch Set 12: Code-Review-1\n\nPlease rebase and fix, would be good to have the new module working in the CI.\n\nAlthough.. it would be also good to have some CLI coverage.","accounts_in_message":[],"_revision_number":12},{"id":"37233b6c405dd6eed515ca3cad8278c616a03211","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-08-25 21:40:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a373b70c5183b69e2e4d4b12b4001ddeff799258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:24:59.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc665c0cae74dceb7b40ad113094bc8 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fee4cce39e342ccb1231eafa72da7c6 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8266d1c4832a40bdb369ea309dba1493 : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c341a69d5fbe48998b0a8187fca63b5b : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a97c6ed7c024d53b1c9c2fbc0272601 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0790ed565347eeac7a26992b2b9c3d : FAILURE in 4m 27s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/7b99cb6d7fff4f4297b560a7b2cddeb8 : FAILURE in 38m 54s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/8859313c1567448b9dc8715141a9cbfe : FAILURE in 32m 47s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/c69c3e4d52ee4130830f058440d73cf5 : FAILURE in 32m 01s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/fc5d520096b04b92921c33f5f46f78ba : FAILURE in 32m 51s","accounts_in_message":[],"_revision_number":13},{"id":"e7f44ee3df06c7d7cb0f5d2f59ccd294316d9863","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-08-26 22:46:20.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ff2ae72dabb6c648be1379da96c8de18c2f8df89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 00:04:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb2d8d8f7ec240b88650b39ab970ac31 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20aa0195f18843cf98ef3d000a99b018 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/074758b96dad4f69a999ff1f9eb63e59 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0881e5a93fbd4e3d8316bae0068cd449 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cbcce54d95041bb94112fe7d7ba935b : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6520466e484ac6af185d64aa11e3e6 : FAILURE in 9m 12s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/69a71257775a4898bc86d9dc9f925f73 : FAILURE in 1h 15m 55s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/03ef5526fd6643dfad800345ba46efa4 : FAILURE in 46m 55s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/45990e3b8a6f427490db33396fce0579 : FAILURE in 47m 02s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/760fda837b23499283ee1a68a8cd5f83 : FAILURE in 57m 08s","accounts_in_message":[],"_revision_number":14},{"id":"ca7f7ebc72b190db9d596a8a2de2332e8d792c59","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-08-27 22:44:23.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"01d48a9c29d7528834fbef3e80bd42dbfcee5185","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-27 23:10:09.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4ecedce19bac459c946587b38959e785 : FAILURE in 24m 05s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b43dc3b4d0cc4b1aa5e1153c48cba3ce : FAILURE in 22m 43s","accounts_in_message":[],"_revision_number":15},{"id":"05f91a69a8f73207e8c9b81b69c3293a8b78ddf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 00:14:00.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9499db4ee1294991a41b2f58bc914183 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41be9bf4885f4a8f8729fe84c43205fc : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e6768e0a8494366ba7d47b59d0bc030 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/285a4b206e464aea90e5cb1a37b6379b : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18c5b3ae6c584e609f675cb80b8c7ab1 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da4b3229e4a450c8660ad846a57bfbf : SUCCESS in 5m 52s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c606adeda2cb44aa98b25129be149db8 : FAILURE in 1h 28m 44s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/7045fcd3d9834ca9bd5b40e2b4c27c72 : FAILURE in 46m 44s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/f53c81d7ea8044abb4c932bd2c5c876c : FAILURE in 50m 23s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/803f85db9406412e8da61289424893f3 : FAILURE in 54m 08s","accounts_in_message":[],"_revision_number":15},{"id":"4fc6ca05791f408b241770d06000f731e07a047b","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-08-28 05:19:27.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"726a8e9f27c5eac8858302c8142efb2d64a33dce","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-08-28 05:41:41.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"ef65e50a356e82624dc40748ef76a33e05dfe5ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-28 06:06:36.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bdf0693c0166416989a7315171cdb390 : FAILURE in 23m 45s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/75e48101ac4a41c49f5a819e7e59a1aa : FAILURE in 24m 08s","accounts_in_message":[],"_revision_number":16},{"id":"14d51c982893dfcaa5fae6c7baf1f174b3ec6f6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 07:34:23.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872d32f0b29a4204ada98924d820fae4 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9d24904ec9647aca25d343be0d0d11b : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc8e3e6a743a4d7f9f111465c6cb1c70 : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afaa550ec6eb4a089182ed513c0271f5 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f37312fb63c49ebab5655874f5ab29f : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6472eb8fc9af4b01aced00b94b255f69 : SUCCESS in 5m 20s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/e33852de953143d5a6ae90af8225449f : SUCCESS in 1h 47m 06s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/96a4259263d4482095dae047eb3cde35 : SUCCESS in 52m 54s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/7882c71ede2445c98d364b66bb37082e : SUCCESS in 1h 02m 21s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/2fdc74d861f14fa8877fc4cf1aa9776b : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":16},{"id":"2dc8fc5141e8d9b1f8b1aa9a35704298e0d684f1","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-01 18:43:11.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"d3b6f98782605a1cdcd1a4f6cf697a41db5a6aba","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-01 23:14:45.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"2f6c9aecbaada1d53b73a5f12ffc054ded01d244","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-01 23:28:32.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"82fa3476cd25a8966868cfd26d451fa00f9b057f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 23:42:14.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3fb2af3b7c49a29fd6d62530933097 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1d11f76624e433b9760af4ea5537c2e : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bf831f86d744ab4aab60fe8d200be9e : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03f92c84fced4ed1b89693290846d7a0 : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a70b497ca294f158d08a60fc678276a : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc59703d82dd472ea8cc5e0e4691f9ab : SUCCESS in 5m 13s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/6e4c267f6abe44309b4694a49bd7d464 : FAILURE in 25m 53s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/7c770e45eb2b4071aa21b8ce41f31b0f : FAILURE in 21m 30s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/832fce4622bd4b6f9632e15d2ddd0583 : FAILURE in 21m 51s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/c8dc272e738147668174626efb6e74c4 : FAILURE in 21m 55s","accounts_in_message":[],"_revision_number":17},{"id":"9d7d1043c7144a971a4c7270ca3f150607596977","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-02 02:00:56.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0ea72060b57841ad8466575dafc2ab37 : SUCCESS in 2h 06m 04s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/77800761a9704432a08b828a00c2998a : SUCCESS in 2h 44m 36s","accounts_in_message":[],"_revision_number":17},{"id":"2c396ba1aaf753018736350972447c9c205d4f22","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-02 16:42:21.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"229e43a35e9f5ead5f5563ebf3a04e18c6789c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 18:35:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42dab35645cd482e8dad2d9d48c0e47d : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c38aec338b104eb0815b2d43c109b495 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4f5fcd9e1d0436eb11fd2d86a17dc78 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07ade44e7ae24679b200625764fce104 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/631d8efeb60840beaf0166ff98ecd455 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e1f8a40d25452790daa5cff6f8a13a : SUCCESS in 8m 03s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/bb82c30ab1f64d2cb47811334eb68eea : SUCCESS in 1h 47m 21s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/147abc2388804484ac8a5d2735ccded6 : SUCCESS in 58m 17s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/fddce19d0f2d4e03bb0f8c77e3e20448 : SUCCESS in 57m 18s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/bd8638eafa7e4420a07468a96e3840e3 : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":17},{"id":"2ba88eaf206ce7c4243ba29bcbe934323d8d5ea9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-02 19:24:47.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a438fd11fb7344e99a16d6094620bac8 : SUCCESS in 2h 23m 27s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c0623b3d7a5c4b56b246a1263ffdaa1e : SUCCESS in 2h 40m 55s","accounts_in_message":[],"_revision_number":17},{"id":"aecf29661148100358df19447753209885ff54c0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-03 09:25:39.000000000","message":"Patch Set 17: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":17},{"id":"93d70ce895503fd367d3535b08271f08049a7dc9","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-03 21:36:01.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"af0aa8bbdbc83f96a9abd19567d64d69456970a9","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-03 22:25:07.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"9aaf93b5d3a26711ace6404eee5899c65084140e","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-04 03:34:15.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"65dc699e7921f9360ae93fd77774bf28ecabd120","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-04 03:34:54.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"abaa1b36e9a7355b76e59d667be84210fb35552d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 05:16:22.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aea1b87b84dd4168bb97511f3381b3b4 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b227cc181154367b06db4215a30b628 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e881132ffc749509535f5c39c35ce26 : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d03190d6327b400f99b3d57802361e00 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6c56d945c0a4e16b35c28e8a4e0ef15 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da69802c1874475b7f76e291d301d7d : SUCCESS in 6m 00s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/12bb0a3cdae244239605f3617de51cea : SUCCESS in 1h 37m 07s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/9c9fad4714254b33855fdaaef2949199 : SUCCESS in 1h 00m 44s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/4c0e9e0c646344be831245afccc429a3 : SUCCESS in 58m 41s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/10ab9133b6a34a9d8b3fb228807f6928 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":18},{"id":"d19aa1bc392b35808e8d7ff843a23a1251d388b7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-04 05:37:41.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e1343a81f6604d6da711a38e207f5ea4 : SUCCESS in 2h 01m 51s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b77ea557877c4e6ab348e20080822f6e : FAILURE in 1h 40m 06s","accounts_in_message":[],"_revision_number":18},{"id":"365052cb8c244fc6aea97fbfe1159609e52d7666","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-06 20:27:36.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"bce5c6304d95f9529eb46b97b4ab3e2e156ed31d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 22:47:18.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b05884d0830848e3873c3b5c340ade81 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0215768c0fd4d4a88fa13e1a06f7047 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90898d4752e14d3cbb32d85c431e2b32 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2448dc4c40da4433843a20f767fd71ef : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46603f6952a84c0dbcd060c7006fda1b : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c73bd5467a4d3888098bfb06efdf3e : SUCCESS in 6m 35s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/ff4aa52d2757435fbbcb34b480f50af1 : SUCCESS in 1h 18m 12s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/58bea27530434f15aa073b5d1e614575 : SUCCESS in 51m 43s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/158b763f298846b69f938e77e293c368 : SUCCESS in 56m 16s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/f0b6b85d36c54340ab528484b170c8b0 : SUCCESS in 52m 40s","accounts_in_message":[],"_revision_number":19},{"id":"b1a25dd3c43ddb7ddcafdecaeaed60cf0fe84217","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-06 22:49:40.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0852ae89a9434576a09db282f79c06aa : SUCCESS in 2h 20m 46s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/3b125dc4e4244eaeb6adabba7bc6485c : SUCCESS in 2h 19m 45s","accounts_in_message":[],"_revision_number":19},{"id":"4dc78bedce60f396d4bc873bcadf66185924d5cb","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-24 01:09:28.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"f2bbc1be9931235216229b9fe46981ef6d3a13b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:27:26.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edaf25c136714222b95220d3fda200c5 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faad44fa1699496089dde94fbdd51895 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b697488adfc4786952c7abb40639993 : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/509c358d2e08436b80b341622287f963 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4bd14f391fe44dfab1781485356eabd : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77fac0250e94f47a4d0375025f5f0c6 : SUCCESS in 7m 06s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/3448c0ffb220475b9d0f6a9a618f84ed : SUCCESS in 1h 12m 38s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/062de5b5bef248eb8cf502e220aea777 : SUCCESS in 52m 45s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/4b3531d89dcb47589601be4839a07e5c : SUCCESS in 54m 58s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/90c302c45a1b498c8c85e9cc72132d02 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":20},{"id":"6e6bce04e5988ea756690e668092c7839a91c465","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-24 03:26:23.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f42d9c93037244f58cdfbf77ccf8f2e8 : SUCCESS in 2h 08m 09s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c99c6dded5d94443adbbc7b07395dd76 : SUCCESS in 2h 14m 49s","accounts_in_message":[],"_revision_number":20},{"id":"2de1094a6a39963da5de53d1ad94914416a98132","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-10-02 14:07:37.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"a3ab78a5204382a2718615f1fa29511b3f74b7d5","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-10-02 14:07:43.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"f5fc1a2925da8121fa1edb9b9fee34798413bea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 15:28:17.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791d4dbadf364a56919d1a236b3e51e0 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b873a358e484bf2b2e680126ba52a3e : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91d57fc94e864b118e90d2b849581de2 : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90847f4f1d7b4403a5adf53246f0cc94 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abee8012b7cb49dab4f0ec480465f0b8 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6719f341504bbb8045c42dd60743a4 : SUCCESS in 7m 56s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/4f5bb270ae5b41f89e7ad4ee6e299351 : SUCCESS in 1h 16m 34s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/83a4437ea912433cabceced3f3456380 : FAILURE in 27m 56s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/47e51dc7b137495694aaba4b892b75cf : SUCCESS in 1h 00m 55s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/59f1f8fe4baa4c4994c7aa6e8b0dd9bf : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":20},{"id":"548d0cbf84b6a4984d9e68efff2753c40d43118b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-02 16:48:54.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e0aedb1325114108972148fcbe8f9bfb : SUCCESS in 2h 10m 18s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/f6736eceffbd4f0981b464ee2502baf8 : SUCCESS in 2h 40m 09s","accounts_in_message":[],"_revision_number":20},{"id":"ee183b46d480c4f91e87b7714b27b9619e35dd1d","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-10-07 20:50:20.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"8ace8d60ddb7452c4561d58bd3b60252c72e387a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 21:57:38.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8672a536885740ed8d099dd13458aae5 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0882b14b9334adbb577582466ec48d1 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/628e1b70a72948feb91567e414c02e79 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87b6b63f087a4e1cb2748e83966fd543 : SUCCESS in 3m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c29b41cfccf94adeaebbb6b5321f43c3 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9502d86eccef4499a9a7fb6b5ef1ab30 : SUCCESS in 6m 59s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/43dc7cb385ad417b974daa61e6ea6dde : SUCCESS in 1h 02m 54s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/77043d6a22924799bebd8788a42ca1da : SUCCESS in 47m 56s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/3ae50642a966440781dd1dc3a7e21c46 : SUCCESS in 46m 03s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/293481ac52b5418f8de3343c545513e8 : SUCCESS in 47m 19s","accounts_in_message":[],"_revision_number":21},{"id":"425843834183ed3f7aee40c28ecd13e2a2d7644d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-07 23:27:28.000000000","message":"Patch Set 21:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c0491bdaf3c64510a9b684830c34981e : SUCCESS in 2h 12m 58s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/e687708036a74cb8a450d2ee6a243a1e : SUCCESS in 2h 35m 49s","accounts_in_message":[],"_revision_number":21},{"id":"e107873b51b4be04d185a26a3fd56120d553559d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-10-08 19:56:38.000000000","message":"Patch Set 21: Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"473031d3b8de798a614cba7582c538dda8f48817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 20:01:47.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"8b10091cdcbcab0d16b4fd11b872a293ffdf64d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:09:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":21},{"id":"eb53faf7d984b7d52fdfef3a49520dc0680ec4f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:09:46.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115a0d0e66974881842b3bd5373ceec7 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b808ac277124f528a44efa2d4de23de : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a53ca13339a4cad8902a09a8546b743 : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9507b9cfd5114aeda8cc2d01e6ff8426 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0425c52bafa144d0a17d9ff84baeb0a3 : SUCCESS in 8m 29s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/aa7a8dfbc6b44499a85433c90865f212 : SUCCESS in 1h 06m 33s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/ce5c3273abb4408bb202d785f5b00276 : SUCCESS in 47m 20s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/84bb8efdd98741dda2d69a9a945a8320 : SUCCESS in 56m 37s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/20921caddc194408961a2b3ae97c6ed6 : SUCCESS in 51m 58s","accounts_in_message":[],"_revision_number":21},{"id":"d4604c590fe769a3b9ca33744187dab4de56beac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:11:33.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a28f7821f3400a9e4cf3881c197420 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"a7bd8fe1ae1ae469f3d19963a0b4be607fe3e40c","revisions":{"b1d3c39a1726291a241a22a327142f4b4cd5a77d":{"kind":"REWORK","_number":1,"created":"2020-06-24 23:13:54.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/1"}}},"commit":{"parents":[{"commit":"d99aec47965c3710651e1753ed3d0694c52b3895","subject":"WIP Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d99aec47965c3710651e1753ed3d0694c52b3895"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 23:13:42.000000000","tz":720},"subject":"WIP metalsmith_deployment role switch to metalsmith_instances","message":"WIP metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nUnresolved differences between the 2 are described below, with\nsuggested solutions:\n\ncandidates is missing from metalsmith_instances. I propose we add a\ncandidates attribute, and also keep the \u0027name\u0027 attribute. If name is\nspecified it is appended to any supplied candidates\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b1d3c39a1726291a241a22a327142f4b4cd5a77d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b1d3c39a1726291a241a22a327142f4b4cd5a77d"}]},"branch":"refs/heads/master"},"55c3713fd9f9c760908a8ee77a878c814b9c942c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-25 03:28:42.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/2"}}},"commit":{"parents":[{"commit":"b7e8e5054ce0ed814f7648b954415e521f3199ee","subject":"WIP Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b7e8e5054ce0ed814f7648b954415e521f3199ee"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-25 03:28:27.000000000","tz":720},"subject":"WIP metalsmith_deployment role switch to metalsmith_instances","message":"WIP metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nUnresolved differences between the 2 are described below, with\nsuggested solutions:\n\ncandidates is missing from metalsmith_instances. I propose we add a\ncandidates attribute, and also keep the \u0027name\u0027 attribute. If name is\nspecified it is appended to any supplied candidates\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/55c3713fd9f9c760908a8ee77a878c814b9c942c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/55c3713fd9f9c760908a8ee77a878c814b9c942c"}]},"branch":"refs/heads/master"},"a4ae6c0c5738049be10b0497b8cbc454130fe08b":{"kind":"REWORK","_number":3,"created":"2020-06-25 23:09:05.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/3"}}},"commit":{"parents":[{"commit":"f5bd0cd140763bf5bbd617b220600e96cac5a6c1","subject":"WIP Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f5bd0cd140763bf5bbd617b220600e96cac5a6c1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-25 23:08:34.000000000","tz":720},"subject":"WIP metalsmith_deployment role switch to metalsmith_instances","message":"WIP metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nUnresolved differences between the 2 are described below, with\nsuggested solutions:\n\ncandidates is missing from metalsmith_instances. I propose we add a\ncandidates attribute, and also keep the \u0027name\u0027 attribute. If name is\nspecified it is appended to any supplied candidates\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a4ae6c0c5738049be10b0497b8cbc454130fe08b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a4ae6c0c5738049be10b0497b8cbc454130fe08b"}]},"branch":"refs/heads/master"},"2f24f17fb3be96f5b3f994aa4c30ac50692e57dd":{"kind":"REWORK","_number":4,"created":"2020-06-26 01:49:45.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/4"}}},"commit":{"parents":[{"commit":"f5bd0cd140763bf5bbd617b220600e96cac5a6c1","subject":"WIP Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f5bd0cd140763bf5bbd617b220600e96cac5a6c1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-26 01:47:51.000000000","tz":720},"subject":"WIP metalsmith_deployment role switch to metalsmith_instances","message":"WIP metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nUnresolved differences between the 2 are described below, with\nsuggested solutions:\n\ncandidates is missing from metalsmith_instances. I propose we add a\ncandidates attribute, and also keep the \u0027name\u0027 attribute. If name is\nspecified it is appended to any supplied candidates\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2f24f17fb3be96f5b3f994aa4c30ac50692e57dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2f24f17fb3be96f5b3f994aa4c30ac50692e57dd"}]},"branch":"refs/heads/master"},"73f36a2363f6384b3473667f909f8653e2395255":{"kind":"REWORK","_number":5,"created":"2020-06-26 03:18:34.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/5"}}},"commit":{"parents":[{"commit":"f5bd0cd140763bf5bbd617b220600e96cac5a6c1","subject":"WIP Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f5bd0cd140763bf5bbd617b220600e96cac5a6c1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-26 03:18:23.000000000","tz":720},"subject":"WIP metalsmith_deployment role switch to metalsmith_instances","message":"WIP metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nUnresolved differences between the 2 are described below, with\nsuggested solutions:\n\ncandidates is missing from metalsmith_instances. I propose we add a\ncandidates attribute, and also keep the \u0027name\u0027 attribute. If name is\nspecified it is appended to any supplied candidates\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/73f36a2363f6384b3473667f909f8653e2395255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/73f36a2363f6384b3473667f909f8653e2395255"}]},"branch":"refs/heads/master"},"b8061a46b8847818979c94fd2fd48a9985a777ba":{"kind":"REWORK","_number":6,"created":"2020-06-28 22:42:48.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/6"}}},"commit":{"parents":[{"commit":"473c801532621a1fc69d9c845853e0d02dfecda8","subject":"Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/473c801532621a1fc69d9c845853e0d02dfecda8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-28 22:42:34.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b8061a46b8847818979c94fd2fd48a9985a777ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b8061a46b8847818979c94fd2fd48a9985a777ba"}]},"branch":"refs/heads/master"},"34b4b32823c960b28cf02bfd757dfa850de29eec":{"kind":"REWORK","_number":7,"created":"2020-06-29 01:36:09.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/7"}}},"commit":{"parents":[{"commit":"473c801532621a1fc69d9c845853e0d02dfecda8","subject":"Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/473c801532621a1fc69d9c845853e0d02dfecda8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-29 01:35:56.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/34b4b32823c960b28cf02bfd757dfa850de29eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/34b4b32823c960b28cf02bfd757dfa850de29eec"}]},"branch":"refs/heads/master"},"3f40252acf7a9683421da846ec1e5026d204d63c":{"kind":"REWORK","_number":8,"created":"2020-06-29 04:30:08.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/8"}}},"commit":{"parents":[{"commit":"473c801532621a1fc69d9c845853e0d02dfecda8","subject":"Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/473c801532621a1fc69d9c845853e0d02dfecda8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-29 04:29:56.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/3f40252acf7a9683421da846ec1e5026d204d63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/3f40252acf7a9683421da846ec1e5026d204d63c"}]},"branch":"refs/heads/master"},"be9d391d22e821035bd4360e7878103d156f2c14":{"kind":"REWORK","_number":9,"created":"2020-07-01 01:38:40.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/9"}}},"commit":{"parents":[{"commit":"62b8b1d8b9697248d4a3368e921b9ee67b73fd0c","subject":"Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/62b8b1d8b9697248d4a3368e921b9ee67b73fd0c"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-07-01 01:38:29.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/be9d391d22e821035bd4360e7878103d156f2c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/be9d391d22e821035bd4360e7878103d156f2c14"}]},"branch":"refs/heads/master"},"7b90dcfc1b7594d8f1178c983651b5b38ddd3215":{"kind":"REWORK","_number":10,"created":"2020-07-01 22:59:24.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/10"}}},"commit":{"parents":[{"commit":"62b8b1d8b9697248d4a3368e921b9ee67b73fd0c","subject":"Move metalsmith_instances from tripleo-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/62b8b1d8b9697248d4a3368e921b9ee67b73fd0c"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-07-01 22:59:02.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7b90dcfc1b7594d8f1178c983651b5b38ddd3215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7b90dcfc1b7594d8f1178c983651b5b38ddd3215"}]},"branch":"refs/heads/master"},"529a24b4bd1e2e97d5ca2827477c6a1bb6bc6a71":{"kind":"REWORK","_number":11,"created":"2020-07-02 02:17:53.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/11"}}},"commit":{"parents":[{"commit":"9361b24c1a7898867f91eb86147de4d216710b5e","subject":"DNM debug image variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/9361b24c1a7898867f91eb86147de4d216710b5e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-07-02 02:16:12.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/529a24b4bd1e2e97d5ca2827477c6a1bb6bc6a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/529a24b4bd1e2e97d5ca2827477c6a1bb6bc6a71"}]},"branch":"refs/heads/master"},"b10a4d606d486d417ae145a02489bfae5091410f":{"kind":"REWORK","_number":12,"created":"2020-07-02 03:54:41.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/12"}}},"commit":{"parents":[{"commit":"9361b24c1a7898867f91eb86147de4d216710b5e","subject":"DNM debug image variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/9361b24c1a7898867f91eb86147de4d216710b5e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-07-02 03:54:24.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b10a4d606d486d417ae145a02489bfae5091410f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b10a4d606d486d417ae145a02489bfae5091410f"}]},"branch":"refs/heads/master"},"5607929a8899adddefb6aac59875c0876036625e":{"kind":"REWORK","_number":13,"created":"2020-08-25 21:40:13.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/13"}}},"commit":{"parents":[{"commit":"f70d0a7c94e15a375926014fb3018f1183e0e848","subject":"Document metalsmith_instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f70d0a7c94e15a375926014fb3018f1183e0e848"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-08-25 21:38:24.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/5607929a8899adddefb6aac59875c0876036625e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/5607929a8899adddefb6aac59875c0876036625e"}]},"branch":"refs/heads/master"},"02f39b72347e0c7141dd6102aee2cec9ab67f171":{"kind":"REWORK","_number":14,"created":"2020-08-26 22:46:20.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/14"}}},"commit":{"parents":[{"commit":"f70d0a7c94e15a375926014fb3018f1183e0e848","subject":"Document metalsmith_instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f70d0a7c94e15a375926014fb3018f1183e0e848"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-08-26 22:40:44.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/02f39b72347e0c7141dd6102aee2cec9ab67f171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/02f39b72347e0c7141dd6102aee2cec9ab67f171"}]},"branch":"refs/heads/master"},"039d4736db77934660ddc1dce0114b9690a911af":{"kind":"REWORK","_number":15,"created":"2020-08-27 22:44:23.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/15"}}},"commit":{"parents":[{"commit":"60ed5399a9c0ec3ae05c8f92c927784a24de1ec8","subject":"Document metalsmith_instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/60ed5399a9c0ec3ae05c8f92c927784a24de1ec8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-08-27 22:32:52.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/039d4736db77934660ddc1dce0114b9690a911af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/039d4736db77934660ddc1dce0114b9690a911af"}]},"branch":"refs/heads/master"},"d9fe8c96601554c4c300b37d2935908323db30ff":{"kind":"REWORK","_number":16,"created":"2020-08-28 05:41:41.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/16"}}},"commit":{"parents":[{"commit":"60ed5399a9c0ec3ae05c8f92c927784a24de1ec8","subject":"Document metalsmith_instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/60ed5399a9c0ec3ae05c8f92c927784a24de1ec8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-08-28 05:41:27.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d9fe8c96601554c4c300b37d2935908323db30ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d9fe8c96601554c4c300b37d2935908323db30ff"}]},"branch":"refs/heads/master"},"88e6a480ef2413eaf8839bbe3e0bfb6360513a73":{"kind":"REWORK","_number":17,"created":"2020-09-01 23:14:45.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/17"}}},"commit":{"parents":[{"commit":"56066a20a476d33c8732279bed74cd8514b198e7","subject":"Document metalsmith_instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/56066a20a476d33c8732279bed74cd8514b198e7"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-09-01 23:13:47.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are diffences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/88e6a480ef2413eaf8839bbe3e0bfb6360513a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/88e6a480ef2413eaf8839bbe3e0bfb6360513a73"}]},"branch":"refs/heads/master"},"14a86cf7103cd59d40aa612be8eaeab864046c90":{"kind":"REWORK","_number":18,"created":"2020-09-04 03:34:54.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/18"}}},"commit":{"parents":[{"commit":"3621535d6733a8493a39eb1a98994694ed8f16c9","subject":"Stop requiring checksums with file images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/3621535d6733a8493a39eb1a98994694ed8f16c9"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-09-04 03:34:40.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are differences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped. A metalsmith_debug arg is added to make\nthe ouput more verbose.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/14a86cf7103cd59d40aa612be8eaeab864046c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/14a86cf7103cd59d40aa612be8eaeab864046c90"}]},"branch":"refs/heads/master"},"dfa8e2d5878747915cb9eccc02906450fa6cafe5":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2020-09-06 20:27:36.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/19"}}},"commit":{"parents":[{"commit":"0bd4b6a6574a51d055a2e2b455e30f29b29aca58","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0bd4b6a6574a51d055a2e2b455e30f29b29aca58"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-09-06 20:26:48.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are differences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped. A metalsmith_debug arg is added to make\nthe ouput more verbose.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/dfa8e2d5878747915cb9eccc02906450fa6cafe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/dfa8e2d5878747915cb9eccc02906450fa6cafe5"}]},"branch":"refs/heads/master"},"0598ba30f4e041d82111c854139d0882f6e985e2":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2020-09-24 01:09:28.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/20"}}},"commit":{"parents":[{"commit":"d3fe5e306be733d98cbb3b3061579e362c6505eb","subject":"Handle out-of-tree openstack module_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d3fe5e306be733d98cbb3b3061579e362c6505eb"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-09-24 01:08:53.000000000","tz":720},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are differences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped. A metalsmith_debug arg is added to make\nthe ouput more verbose.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0598ba30f4e041d82111c854139d0882f6e985e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0598ba30f4e041d82111c854139d0882f6e985e2"}]},"branch":"refs/heads/master"},"a7bd8fe1ae1ae469f3d19963a0b4be607fe3e40c":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2020-10-07 20:50:20.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/88/737888/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/88/737888/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/88/737888/21"}}},"commit":{"parents":[{"commit":"4de71152472d3fafe1b4b8d76ea3369323d620fd","subject":"Handle out-of-tree openstack module_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/4de71152472d3fafe1b4b8d76ea3369323d620fd"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-06-24 22:55:27.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-10-07 20:48:20.000000000","tz":780},"subject":"metalsmith_deployment role switch to metalsmith_instances","message":"metalsmith_deployment role switch to metalsmith_instances\n\nInstead of wrapping the metalsmith CLI, the metalsmith_deployment role\nnow uses the metalsmith_instances module. There are differences between\nthe instances formats of the role and the module which are partially\nresolved with a simple transformation module called\nmetalsmith_deployment_defaults.\n\nA \u0027candidates\u0027 attribute is added to metalsmith_instances, but keeping\nthe single \u0027name\u0027 attribute to remain compatible with the TripleO\nusage.\n\nUnresolved differences between the 2 are described below:\n\nmetalsmith_instances doesn\u0027t have a per-instance state attribute,\ninstead it has one state attribute for all instances. I propose that\nsupport for per-instance state is dropped. This was never documented\nin the README.rst anyway.\n\nextra_args only applies to a CLI. Apart from --dry-run these arguments\nare either for output formatting or Ironic API authentication. I propose\nthat this option is dropped. A metalsmith_debug arg is added to make\nthe ouput more verbose.\n\nChange-Id: Ia30620821182c58050813e807cdde50a27d03c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a7bd8fe1ae1ae469f3d19963a0b4be607fe3e40c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a7bd8fe1ae1ae469f3d19963a0b4be607fe3e40c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
