)]}'
{"id":"openstack%2Fmetalsmith~749719","triplet_id":"openstack%2Fmetalsmith~master~Ia123c1d3c57cc2814e3f971209cbee3ab336f7bd","project":"openstack/metalsmith","branch":"master","hashtags":[],"change_id":"Ia123c1d3c57cc2814e3f971209cbee3ab336f7bd","subject":"Stop requiring checksums with file images","status":"MERGED","created":"2020-09-03 12:24:52.000000000","updated":"2020-09-03 15:50:31.000000000","submitted":"2020-09-03 15:48:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749719-1599148137567-1189cc01","meta_rev_id":"3d4f718c8c4901e1d5fa695a8df75045653bf308","_number":749719,"virtual_id_number":749719,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 15:48:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-09-03 12:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-03 13:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2020-09-03 14:16:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 12:28:13.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-09-03 13:22:29.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-09-03 15:48:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"235f300780c84893301468f4c2b20ef0e661ceb6","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-03 12:24:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3697c4d848270e35f5eec924b2cb9bcb318c0441","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-09-03 12:28:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"21cf24a977e090ccd035f5ba7bf820c35bebca42","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-09-03 13:22:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"020ea50bf0f5cb26d5bcf23782c9beda070243cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 14:12:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16f2f8479b1c48f6857e65a696dbd520 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9db854322c4740d5a7ab0e2872050029 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4ed776c9b49473989c4938071c16679 : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb2206c772b2487498ac5ebd75054578 : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fff7395e3620402ab3d2a6766e85981e : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b945e1d6d10c4cbe94f4c958521725d2 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5a08229e2c9438dabe848a2c1e30470 : SUCCESS in 3m 50s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c4373350215c48279f571698c29c2192 : SUCCESS in 1h 40m 41s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/3d1cfcd3a96145d4b1a295b744a45ca0 : SUCCESS in 59m 17s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/a49ef7644e4a4bb79bf90e9b6fbb8386 : SUCCESS in 55m 16s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/6fbfd5ad17bc4fedb8dd047ff01079ed : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7308e917e89cda0ff148f0e9cb3dd286c09f24c2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-03 14:16:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d4cceaca672f5b0364836514a7c5f0e9886850a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 14:16:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec224094a389ca16f34ea130f249679bf7bc4db4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 15:48:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e328483d8b436b827b3dfb1c2b5832 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d446965a034441fe9dd651545c601865 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eedd54a410f04d2ab60cb29696886700 : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97735a2956b24309a51daba7dd422732 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f7015c58704a7c9fa94ae56879c432 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bcc1a5f6335429486a676c89eed588a : SUCCESS in 4m 22s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/2fa11058566f47df8d380f3c6945718e : SUCCESS in 1h 28m 08s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/5bc7b046d30440e9aa781ea4d669ba07 : SUCCESS in 1h 03m 06s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/8f7d373cae46497dac84c0730845a57a : SUCCESS in 56m 55s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/023ea348df43497f95d67a94ba0d26fe : SUCCESS in 47m 25s","accounts_in_message":[],"_revision_number":1},{"id":"95b54669286296571f94487a2dc135f28932bede","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 15:48:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d4f718c8c4901e1d5fa695a8df75045653bf308","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 15:50:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e0be3261a4844298ea319f5455e5e17 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/102698dbea364735ba57332a9fa39e33 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3621535d6733a8493a39eb1a98994694ed8f16c9","revisions":{"3621535d6733a8493a39eb1a98994694ed8f16c9":{"kind":"REWORK","_number":1,"created":"2020-09-03 12:24:52.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/19/749719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/19/749719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/19/749719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/19/749719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/19/749719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/19/749719/1"}}},"commit":{"parents":[{"commit":"97099bc0fffd6886694638842549151046a5dadd","subject":"Merge \"Document metalsmith_instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/97099bc0fffd6886694638842549151046a5dadd"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-09-03 12:23:52.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-09-03 12:23:52.000000000","tz":120},"subject":"Stop requiring checksums with file images","message":"Stop requiring checksums with file images\n\nNeither deploy method requires checksums with file images, they\u0027re\nsimply ignored. Deprecate providing them.\n\nChange-Id: Ia123c1d3c57cc2814e3f971209cbee3ab336f7bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/3621535d6733a8493a39eb1a98994694ed8f16c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/3621535d6733a8493a39eb1a98994694ed8f16c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
