)]}'
{"id":"openstack%2Fmetalsmith~750489","triplet_id":"openstack%2Fmetalsmith~master~I411e2bc9c15ee4fb77fba980e1924823fe16ba3f","project":"openstack/metalsmith","branch":"master","hashtags":[],"change_id":"I411e2bc9c15ee4fb77fba980e1924823fe16ba3f","subject":"Change metalsmith nodeset to Focal","status":"MERGED","created":"2020-09-08 20:37:36.000000000","updated":"2020-09-18 12:05:30.000000000","submitted":"2020-09-18 11:55:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750489-1600430105144-7e3b2230","meta_rev_id":"23901fb392b6526e2b45e0f5c8cbc9e5f1a9b00b","_number":750489,"virtual_id_number":750489,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 11:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"recommended":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":1,"date":"2020-09-16 22:53:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-10 15:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":2,"date":"2020-09-18 10:13:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":1,"date":"2020-09-18 10:13:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 15:14:36.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-09-16 22:53:47.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2020-09-18 10:13:35.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-09-18 11:55:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef1757c26973d36850e8f28b0faa7221a71b5b53","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-08 20:37:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"925ae1a6bbffc9678236e6e6ce6a33433913d904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:17:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fcea9cbb69a431891482fedd2177cc8 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06a34d78f3474b8ebc66bf7e1b4e3b5b : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c7d4403b3854c729ab19a991366a1de : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcf0383e8e554241803e2224a2dbdbc6 : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3c0401af63c4931bcdd3d4ae9931054 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25a89fb67514c748af6d4f60b77efbd : SUCCESS in 8m 06s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/ce5bd2d187ee493e8693e7c7d239c3a2 : FAILURE in 24m 30s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/09d0aa81e2044090a2ae0b26c048d0b2 : FAILURE in 30m 34s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/f6c85ddae8524a348f39ccaf46ad6d4e : FAILURE in 33m 24s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/cca07547be6c48cfb457f9be971c074e : FAILURE in 33m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a69a1342f130cfc3435afe2f76396c4c52bc8caa","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-08 22:04:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"908c3cae074c9579bc9221441a114eed5950a41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 23:16:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19560a21920247ffb10d5d6cfeee4871 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c325848126a849c99b7214d66c1c25ed : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff443802d0b44740ba16b9216ad5dcc5 : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4536753712c14b27809cf08537f9916d : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13646d525e77464786e6021fcc84966a : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902837131ed84431afb83087341fbd88 : SUCCESS in 7m 38s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c6ce65543ac14bababf51b50e94ab6e9 : SUCCESS in 58m 38s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/40ff9c51ecdc4fe98fb6bd7d0a005e41 : SUCCESS in 1h 00m 06s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/6f1cd33125984cffa6b5fec4ac07d730 : SUCCESS in 54m 58s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/1072a25ce35c4f9593562e57732257b5 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":2},{"id":"7822c7a7ddaf2c03ba62a9f8755823c7d338e682","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-09 18:35:24.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"140e4cc74156330f810d3ea2110da04f75cc4e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 19:49:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd5dd43be6d4f4ba88c1e7ac0639052 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d150197a769a4f88ac1b878f39d966c3 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67fee58492934a8e825ffe57e5820356 : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eea7d4d5a93141aeb64456ba93d2772e : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eadaa7a394c24190a66a5764176b711c : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242e4d53041e49559f339add9ac9dafd : SUCCESS in 6m 34s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/6a34938aac484a048738e2cbbba38d92 : SUCCESS in 1h 09m 08s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/db1e6f66ecb649b59d2d2e7bd90a7138 : SUCCESS in 53m 11s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/b08f64f24fc04b4f9eea739ae30c5d01 : SUCCESS in 55m 05s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/46c771b009fc40cfa1815cded50171ad : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":3},{"id":"e9b21a4bcfe74dc762b0cab87325b69aca7599ad","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-10 15:14:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0eda8b1045d9cb75e8f5b3f65b8dd896177bc221","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-09-16 22:53:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0f848031b0f73179ec0fad9bed76945675a22e81","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-09-18 10:13:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9cb24c66bdb2e5a27d52b61eff587126f3929248","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 10:14:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b62c3df90bc29fbccfd5c51b0d26bf72d8eb47e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:55:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e984cdfcafc04e739845be3b672c0da5 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f4d239c7b5344feb961f36aae87ed85 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37ed9661b92b41108e8fa6613ae6b23d : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92fec90263b2468cabd5ed74f7999f24 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45c9e2c7c8c4dd493c3a1c49fd43086 : SUCCESS in 15m 01s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/cfc5e7d50d0d4fbd8aa54d2b84b54f16 : SUCCESS in 1h 03m 17s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/02bddbd42bd54a049b146c81bafdb611 : SUCCESS in 51m 50s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/d890b6844b3d4c18b35f2c76a2b4743e : SUCCESS in 53m 40s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/a5448b0557ff4e87b25a47c8fe80f1a2 : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":3},{"id":"bf1666ca82ae2084718ea71875ebb06e9347d0b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:55:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"23901fb392b6526e2b45e0f5c8cbc9e5f1a9b00b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 12:05:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d2ee6b267a4d4fad7128f32b6ac29c : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e375251c9469104d1386398d4ac270a9d6fac73c","revisions":{"d09ddcc2da26e142f4a4cbfd5011eaa04dda8753":{"kind":"REWORK","_number":1,"created":"2020-09-08 20:37:36.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/89/750489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/89/750489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/1"}}},"commit":{"parents":[{"commit":"a42880c79f3264ca1511fd9cce34bf528ffcad6e","subject":"Rework jobs so failed devstack does not retry the job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a42880c79f3264ca1511fd9cce34bf528ffcad6e"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-09-08 20:36:16.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-09-08 20:36:16.000000000","tz":-420},"subject":"Change metalsmith nodeset to Focal","message":"Change metalsmith nodeset to Focal\n\nBy default, metalsmith jobs relied upon the devstack\ndefault which would pull in bionic, but as we want to\ntest UEFI soon, we need to have focal jobs for the working\nuefi driver PXE network stack.\n\nChange-Id: I411e2bc9c15ee4fb77fba980e1924823fe16ba3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d09ddcc2da26e142f4a4cbfd5011eaa04dda8753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d09ddcc2da26e142f4a4cbfd5011eaa04dda8753"}]},"branch":"refs/heads/master"},"ebb23fa2e70b76e0c94a13d57c57810548649bd8":{"kind":"REWORK","_number":2,"created":"2020-09-08 22:04:32.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/89/750489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/89/750489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/2"}}},"commit":{"parents":[{"commit":"a42880c79f3264ca1511fd9cce34bf528ffcad6e","subject":"Rework jobs so failed devstack does not retry the job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a42880c79f3264ca1511fd9cce34bf528ffcad6e"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-09-08 20:36:16.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-09-08 22:03:50.000000000","tz":-420},"subject":"Change metalsmith nodeset to Focal","message":"Change metalsmith nodeset to Focal\n\nBy default, metalsmith jobs relied upon the devstack\ndefault which would pull in bionic, but as we want to\ntest UEFI soon, we need to have focal jobs for the working\nuefi driver PXE network stack.\n\nAlso disables dstat as we don\u0027t presently use it and it seems\nbroken on focal. :\\\n\nChange-Id: I411e2bc9c15ee4fb77fba980e1924823fe16ba3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ebb23fa2e70b76e0c94a13d57c57810548649bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ebb23fa2e70b76e0c94a13d57c57810548649bd8"}]},"branch":"refs/heads/master"},"e375251c9469104d1386398d4ac270a9d6fac73c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-09 18:35:24.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/89/750489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/89/750489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/89/750489/3"}}},"commit":{"parents":[{"commit":"e0ed4d2dee5028b0ea0bfc7f27720cf24106b065","subject":"Rework jobs so failed devstack does not retry the job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e0ed4d2dee5028b0ea0bfc7f27720cf24106b065"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-09-08 20:36:16.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-09-09 18:35:24.000000000","tz":0},"subject":"Change metalsmith nodeset to Focal","message":"Change metalsmith nodeset to Focal\n\nBy default, metalsmith jobs relied upon the devstack\ndefault which would pull in bionic, but as we want to\ntest UEFI soon, we need to have focal jobs for the working\nuefi driver PXE network stack.\n\nAlso disables dstat as we don\u0027t presently use it and it seems\nbroken on focal. :\\\n\nChange-Id: I411e2bc9c15ee4fb77fba980e1924823fe16ba3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e375251c9469104d1386398d4ac270a9d6fac73c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e375251c9469104d1386398d4ac270a9d6fac73c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
