)]}'
{"id":"openstack%2Fmetalsmith~767978","triplet_id":"openstack%2Fmetalsmith~stable%2Fvictoria~I52ccba571669253010c6c12ede722ce314d9c12b","project":"openstack/metalsmith","branch":"stable/victoria","topic":"remove-l-c-job","hashtags":[],"change_id":"I52ccba571669253010c6c12ede722ce314d9c12b","subject":"Remove lower-constraints job","status":"MERGED","created":"2020-12-19 14:41:58.000000000","updated":"2020-12-19 20:14:32.000000000","submitted":"2020-12-19 20:10:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767978-remove-l-c-job","cherry_pick_of_change":767090,"cherry_pick_of_patch_set":2,"meta_rev_id":"ebabf163876877516eaf74d6b6fa030f3194ba07","_number":767978,"virtual_id_number":767978,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-19 20:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-19 19:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-19 19:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-19 14:42:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-19 19:06:36.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"5d3981a4169d37a433dbec4a353f5fb7a8d06310","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-19 14:41:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"51840a34b857449be0d14fe6dac08567538c6e73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 14:42:17.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/metalsmith on branch stable/victoria.  The error was:\n\n  while scanning a simple key\n  in \"openstack/metalsmith/.zuul.yaml@stable/victoria\", line 177, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/metalsmith/.zuul.yaml@stable/victoria\", line 180, column 1","accounts_in_message":[],"_revision_number":1},{"id":"d31984de1fb022317b1ded616398eb32201a4215","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-19 14:42:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f3facf963538e8c0ddb379aa9d2edf7adf5657f8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-19 14:42:39.000000000","message":"Topic remove-l-c-job-stable/victoria removed","accounts_in_message":[],"_revision_number":2},{"id":"a062dbadb4b15e48961a5f53d911686ce0f316b8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-19 14:42:45.000000000","message":"Topic set to remove-l-c-job","accounts_in_message":[],"_revision_number":2},{"id":"449752b700c1a992314f4f5e08139c11a28bc74e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-19 14:42:47.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"3bc98bfb810013964190068c1db5765c513faed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 15:50:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59261cd209c24402a8670cef6b272441 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b88a51da5a646bf829c5a283ba9573f : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3b66d1462bb4937aa001b8ce5ec2cd0 : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c7e03a349ba4ac3afa5ddaf43a732d5 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/650fb661d96346b091113925472e325c : SUCCESS in 6m 50s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c6267bc25d574f69b9589aeacf3a5ddc : SUCCESS in 1h 06m 15s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/9d0d2d63c5904bc08b2090c2d03f68dd : SUCCESS in 41m 43s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/d6c33bbb8dc54d8f965fa4a8410bc7ad : SUCCESS in 44m 52s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/26e2a7c453cf4f2291c8534b1fa03228 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":2},{"id":"7a3f0772a5f347c367b183190168b15271297a85","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-19 19:06:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70bd0fe0ce68cddcebf5c478107f696dc4382339","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 19:06:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7492a1419bb0908f0032d08907a46a656f8eb9ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 20:10:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a052c0d3941940b58a9f241d1aef2381 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c01e26e57894c2a9a147f2544e9328c : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8f1ee635fe44807a17cc6930a7fdadd : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43d10de9fec45c5a480595f4ffd3516 : SUCCESS in 7m 23s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/356293e2383c4273851233b51ad2afdc : SUCCESS in 1h 01m 10s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/1aa61554b2934b05b5e9b2f027364346 : SUCCESS in 51m 09s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/995dd886c56c454e8d904e3ad6b6b97a : SUCCESS in 52m 20s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/485e8a1a69c94bbba005e19b6716dedb : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7d81b295a13f55b6532d75cb4472b39e03e2c6e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 20:10:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ebabf163876877516eaf74d6b6fa030f3194ba07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 20:14:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6471ba00848342f38ba63521eeee43ba : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1cb59379a5fe64eb1eeb8c40178f4263d2fa535","revisions":{"12f7bade3f2d8b84e71875038c5767d80cd14827":{"kind":"REWORK","_number":1,"created":"2020-12-19 14:41:58.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/78/767978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/78/767978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/78/767978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/78/767978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/78/767978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/78/767978/1"}}},"commit":{"parents":[{"commit":"fc2b3f39830803019f1eef01c22303a4486b0bf1","subject":"Merge \"Rework jobs so failed devstack does not retry the job\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/fc2b3f39830803019f1eef01c22303a4486b0bf1"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-15 08:20:45.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-19 14:41:58.000000000","tz":0},"subject":"Remove lower-constraints job","message":"Remove lower-constraints job\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nChange-Id: I52ccba571669253010c6c12ede722ce314d9c12b\n(cherry picked from commit 33c7e7feedd025d3603944950164d6b2a098f0bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/12f7bade3f2d8b84e71875038c5767d80cd14827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/12f7bade3f2d8b84e71875038c5767d80cd14827"}]},"branch":"refs/heads/stable/victoria"},"d1cb59379a5fe64eb1eeb8c40178f4263d2fa535":{"kind":"REWORK","_number":2,"created":"2020-12-19 14:42:35.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/78/767978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/78/767978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/78/767978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/78/767978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/78/767978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/78/767978/2"}}},"commit":{"parents":[{"commit":"fc2b3f39830803019f1eef01c22303a4486b0bf1","subject":"Merge \"Rework jobs so failed devstack does not retry the job\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/fc2b3f39830803019f1eef01c22303a4486b0bf1"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-15 08:20:45.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-19 14:42:33.000000000","tz":0},"subject":"Remove lower-constraints job","message":"Remove lower-constraints job\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nChange-Id: I52ccba571669253010c6c12ede722ce314d9c12b\n(cherry picked from commit 33c7e7feedd025d3603944950164d6b2a098f0bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d1cb59379a5fe64eb1eeb8c40178f4263d2fa535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d1cb59379a5fe64eb1eeb8c40178f4263d2fa535"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
