)]}'
{"id":"openstack%2Fmetalsmith~813400","triplet_id":"openstack%2Fmetalsmith~master~If5c8feb3a829eaff56519846de3e5ccd6bab24c4","project":"openstack/metalsmith","branch":"master","topic":"ironic-default-uefi","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2021-10-11 16:48:31.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"If5c8feb3a829eaff56519846de3e5ccd6bab24c4","subject":"Set explicit boot mode to bios for legacy job","status":"MERGED","created":"2021-10-11 09:51:12.000000000","updated":"2021-10-11 16:54:21.000000000","submitted":"2021-10-11 16:48:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813400-ironic-default-uefi","meta_rev_id":"c3a8a1c1841b476ce5bc3df56e654894f18264cc","_number":813400,"virtual_id_number":813400,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 16:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-11 15:25:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2021-10-11 15:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2021-10-11 15:27:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 11:32:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 15:25:39.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-10-11 15:27:08.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"ce77be3409d595ff517b40880c77992800e669b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-11 09:51:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a2dc56d61e4d3d17c34c42851cbd559d508568d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 11:32:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/650b385b9b1b45d8b4b1d54e174bcf3a : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bc1b9b090284d2aaa70ab751026e9bb : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bd62f6d2396402698476b697717f877 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64551c2f9c194254892bebbcb051ad5c : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bb1434ba3484921b523cbc66c2ac175 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6f5b627c194fcdb68ae20e815f2b57 : SUCCESS in 8m 05s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/b714c5f3ce3149669b22f009dece753d : SUCCESS in 1h 34m 27s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/6550b2fa02c04a07aba36e6853729ee6 : FAILURE in 45m 36s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/d7254e6ad76341a58b7c00b3d1ce3e28 : FAILURE in 50m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6c238102ae5a9b116b3d2a7228cea81994333269","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-11 12:47:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7558e7c0703fb4c2da1cb0dad6fc2bec0fe62a42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 14:27:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba91f7d95e5489a8a6ff1830db2d44e : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfad9ef5e4da4edf855cbe5ef368b507 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26110b9026ed4836af4f5f77f30a2c91 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a51477d5ea5444fa9bd587104bfb6ceb : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5605574ab7240aa80cd5aab5d441876 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4eb02f202346e68854324edd13ad28 : SUCCESS in 8m 08s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/8edbcbf7a3e64f209fc4cf0ef4694c19 : SUCCESS in 1h 35m 39s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/d243cd2b55e24ed59a8f301855f07fd2 : SUCCESS in 1h 10m 57s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/6130b3643a6c444fb857b5155ed4867f : SUCCESS in 48m 30s","accounts_in_message":[],"_revision_number":2},{"id":"15cc2c564f29f70b1e13fd3da0242999ea2eb2c3","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-11 15:25:31.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"b722207976c810fc02c8b4596bb4d16ecc1a0ad9","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-10-11 15:25:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2a6e74e879db8fc5d859742ca6932fba33fa9d5b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-10-11 15:27:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81d9ee43f7a11fb3234da147e876a3f8f8fbd362","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 15:28:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c3a9963ed22e01414aa441c19060cfe5ac6ad30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:48:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7aec5aa0654fe3bd0eee4d56196337 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31f93347b80d4463a1e6ff9c4d1b9530 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0226a80e3924e42b0de269857cdc777 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a8bdcb624540129c0bf37afbf57f73 : SUCCESS in 7m 53s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/0ccd5d69edc44bf49af69badc4f2f49e : SUCCESS in 1h 18m 33s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/3e0fe375c47b4e96a5e00735bf9e51e3 : SUCCESS in 52m 45s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/45c0a7205674460e9c79ffa90a4f5978 : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d46aac6d40ada6b654fa34fe7cc1c243d0d66ffc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:48:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c3a8a1c1841b476ce5bc3df56e654894f18264cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:54:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0169f2ad8f4e4afa9e96519aade3ea42 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab69212fe72bd3004513f7eaba7f36769aad8b41","revisions":{"ab7ec9a7762fb830099902066b18deb459508fbc":{"kind":"REWORK","_number":1,"created":"2021-10-11 09:51:12.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/813400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/00/813400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/813400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/813400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/813400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/00/813400/1"}}},"commit":{"parents":[{"commit":"f98dfa61c1d7475b81c20dabbf2c74198c38c793","subject":"Write network_data.json metadata to config-dirve","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f98dfa61c1d7475b81c20dabbf2c74198c38c793"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-10-11 09:49:50.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-10-11 09:49:50.000000000","tz":120},"subject":"Set explicit boot mode to bios for legacy job","message":"Set explicit boot mode to bios for legacy job\n\nThe default boot mode for ironic is now uefi [1]\nThe legacy jobs should set IRONIC_BOOT_MODE explicitely to bios.\n\n[1] https://opendev.org/openstack/ironic/commit/493b4f0caf591caeaa3285c7f597a54121d99985\n\nChange-Id: If5c8feb3a829eaff56519846de3e5ccd6bab24c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ab7ec9a7762fb830099902066b18deb459508fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ab7ec9a7762fb830099902066b18deb459508fbc"}]},"branch":"refs/heads/master"},"ab69212fe72bd3004513f7eaba7f36769aad8b41":{"kind":"REWORK","_number":2,"created":"2021-10-11 12:47:03.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/813400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/00/813400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/813400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/813400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/00/813400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/00/813400/2"}}},"commit":{"parents":[{"commit":"f98dfa61c1d7475b81c20dabbf2c74198c38c793","subject":"Write network_data.json metadata to config-dirve","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f98dfa61c1d7475b81c20dabbf2c74198c38c793"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-10-11 09:49:50.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-10-11 12:46:57.000000000","tz":120},"subject":"Set explicit boot mode to bios for legacy job","message":"Set explicit boot mode to bios for legacy job\n\nThe default boot mode for ironic is now uefi [1]\nThe legacy jobs should set IRONIC_BOOT_MODE explicitely to bios.\n\n[1] https://opendev.org/openstack/ironic/commit/493b4f0caf591caeaa3285c7f597a54121d99985\n\nChange-Id: If5c8feb3a829eaff56519846de3e5ccd6bab24c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ab69212fe72bd3004513f7eaba7f36769aad8b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ab69212fe72bd3004513f7eaba7f36769aad8b41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
