)]}'
{"id":"openstack%2Fmetalsmith~821242","triplet_id":"openstack%2Fmetalsmith~master~I5c7928e341ffcbc94f1beb8a33a8d10d89bae8a7","project":"openstack/metalsmith","branch":"master","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2021-12-09 21:08:44.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I5c7928e341ffcbc94f1beb8a33a8d10d89bae8a7","subject":"CI: Extend metalsmith CI job timeout","status":"MERGED","created":"2021-12-09 16:41:14.000000000","updated":"2021-12-09 23:07:44.000000000","submitted":"2021-12-09 21:08:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821242","meta_rev_id":"716ff503c1310e037785bcbc75e50ae35359e280","_number":821242,"virtual_id_number":821242,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-09 21:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":2,"date":"2021-12-09 19:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":1,"date":"2021-12-09 19:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 18:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-09 19:34:23.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"ace5cd1b3ce9ff3e04c2bf7579dd237f5e761854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-12-09 16:41:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09ed144baef710b3d81837dfc0daae29393c96ab","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-12-09 16:43:19.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"5dea0f8436763219755f6eef51fcf534f9fbc36e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 18:02:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c123098cee471ea88bce9f77e6125e : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/850eba3a7ae64bdc8c075e07e36d964f : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f504b6ddc2fb498fad60a39ce3c2b4bb : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28b4e0852b234ef88f5c1af2e5d8256c : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2adbc0fdd24ef6a4a34b1e68c18546 : SUCCESS in 8m 05s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/e35a9212e29f45dca878628f04dcd795 : SUCCESS in 1h 18m 44s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/11f0bc96fb9846ed9f8df78c1b37437f : SUCCESS in 49m 45s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/2c7edb67c80f45e9823f50bd1fc9cc19 : SUCCESS in 48m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c6a743c1ee42f14365afa397a7c75ba4b2713a9d","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-12-09 19:34:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70950917827e2fc42a32c1954eaa8214e29685a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 19:35:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08e56a51f0cb1604666a1faff2bf1996b1381f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 21:08:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c97975c08ff4c89b5f90a2c5a16b132 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4823458ac2e455f84b6117c5e08d0d1 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79e65a860e544bbebcfa95b438e9a38d : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0dd3c15e4040968f2ce507bc2c52a3 : SUCCESS in 8m 14s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/fea13c4e7f574eb2af6a429122b6f7fc : SUCCESS in 1h 19m 40s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/228d4de2bea84dec9902878286104e52 : SUCCESS in 51m 36s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/469a08c6833a4b9392ac2fc5ddbc46df : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2eed58bca7186df974a1592ef4754933c756195a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 21:08:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"716ff503c1310e037785bcbc75e50ae35359e280","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 23:07:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775f64cc24ef4170950daf97fab81ebc : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"347c354d8d50ec96ed191a725bcbff32a8591f00","revisions":{"347c354d8d50ec96ed191a725bcbff32a8591f00":{"kind":"REWORK","_number":1,"created":"2021-12-09 16:41:14.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/42/821242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/42/821242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/42/821242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/42/821242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/42/821242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/42/821242/1"}}},"commit":{"parents":[{"commit":"6c830e048f470fdcbd22c64a1cf69f8f8a59e7be","subject":"Use only Yoga tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/6c830e048f470fdcbd22c64a1cf69f8f8a59e7be"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-12-09 16:38:19.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-12-09 16:38:19.000000000","tz":-480},"subject":"CI: Extend metalsmith CI job timeout","message":"CI: Extend metalsmith CI job timeout\n\nMetalsmith has always been a fairly long running job, in part\nbecause it hast to build images, and then deploy them.\n\nIt appears, we are seeing increased occurances of the job\ntiming out due to what appears to be just slightly slower IO\nperformance. Overall major steps are still executing in an\nacceptable amount of time, just towards the further end of\nwhat we typically expect to see in CI.\n\nAs such, extend the overall job timeout a little bit to give\nit time to hopefully finish. Extends the job timeout 1000 seconds,\nwhich is a bit over sixteen more minutes to allow the job to\ncomplete cleaning.\n\nChange-Id: I5c7928e341ffcbc94f1beb8a33a8d10d89bae8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/347c354d8d50ec96ed191a725bcbff32a8591f00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/347c354d8d50ec96ed191a725bcbff32a8591f00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
