)]}'
{"id":"openstack%2Fmetalsmith~821665","triplet_id":"openstack%2Fmetalsmith~master~Id8d0e53ddf18b250388ea508898af8a0645b6e19","project":"openstack/metalsmith","branch":"master","topic":"yoga-python-testing","hashtags":[],"change_id":"Id8d0e53ddf18b250388ea508898af8a0645b6e19","subject":"Re-add python 3.6/3.7 in classifier","status":"ABANDONED","created":"2021-12-14 08:27:34.000000000","updated":"2021-12-14 08:39:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"45d6820640f6de1c9448aeb47187b6f7455611e9","_number":821665,"virtual_id_number":821665,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-14 08:37:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-14 08:37:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22e618c19081171969fe7f1ffe9dedf1fb31f656","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-12-14 08:27:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a3dead846556e09fd3f39bfffa0912c3cd12ae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 08:37:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cffca9b328b2414f832a4e7fc9b2702e : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce16d56d7cb47f7a9d01b3b19dc2651 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c39d45ca5eba497bb1e5b50eb4dbabcf : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7a862e7daa3417c846bc3c966fe557b : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b666c7cc0bbf451f970554ed6703c288 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d44a0b8f703487193dac47462275c32 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"45d6820640f6de1c9448aeb47187b6f7455611e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-12-14 08:39:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2baee22001426cc3e69ada6b1a73b7339e3fe9f1","revisions":{"2baee22001426cc3e69ada6b1a73b7339e3fe9f1":{"kind":"REWORK","_number":1,"created":"2021-12-14 08:27:34.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/65/821665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/65/821665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/65/821665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/65/821665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/65/821665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/65/821665/1"}}},"commit":{"parents":[{"commit":"347c354d8d50ec96ed191a725bcbff32a8591f00","subject":"CI: Extend metalsmith CI job timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/347c354d8d50ec96ed191a725bcbff32a8591f00"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-12-14 08:27:04.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-12-14 08:27:19.000000000","tz":60},"subject":"Re-add python 3.6/3.7 in classifier","message":"Re-add python 3.6/3.7 in classifier\n\nWe have updated the yoga testing runtime to keep the\npy36 testing.\n\n- https://review.opendev.org/c/openstack/governance/+/820195\n\nUnit tests job template is also updated to keep python\n3.6 as a voting job. So with the py3.6 and py3.9 testing as voting\njob template, we are keeping python 3.6, 3.7, 3.8, and 3.9 as\ntested versions in the Yoga cycle.\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nThis commit re-add the python 3.6/3.7 versions in setup.cfg classifier.\n\nChange-Id: Id8d0e53ddf18b250388ea508898af8a0645b6e19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2baee22001426cc3e69ada6b1a73b7339e3fe9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2baee22001426cc3e69ada6b1a73b7339e3fe9f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
