)]}'
{"id":"openstack%2Fmetalsmith~835749","triplet_id":"openstack%2Fmetalsmith~master~I2c772b2d26bd7922dd9790a6edbeab76fcb1b62f","project":"openstack/metalsmith","branch":"master","topic":"move_to_cs9","attention_set":{},"removed_from_attention_set":{"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2022-04-19 11:51:18.000000000","reason":"Iury Gregory replied on the change"}},"hashtags":[],"change_id":"I2c772b2d26bd7922dd9790a6edbeab76fcb1b62f","subject":"Move from CS8 to CS9","status":"ABANDONED","created":"2022-03-29 19:23:30.000000000","updated":"2022-04-19 11:51:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d576db53145eb39ef17302d5b316c8d6e8f80bc2","_number":835749,"virtual_id_number":835749,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-29 20:56:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 20:56:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 10:56:29.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"}],"messages":[{"id":"abd2ed5686002baf6e758936bb98e992d2b17aec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-03-29 19:23:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58b0daf701beecad84e6eee402aefb21de39b8ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 20:56:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b397d6957dd4fce837473245b9e5120 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1f15ad9a2d4657bfb1b19ccb616e0f : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e194a4e5c71c4d2fa05e71e429b5f9ac : SUCCESS in 8m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7fa9826dc134f76885950d3653cac75 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8219b19dfc32450bb11b3e638c844450 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e34e59988d7b4758b8a160938891094f : SUCCESS in 11m 00s\n- metalsmith-integration-glance-localboot-centos9-uefi https://zuul.opendev.org/t/openstack/build/e8bae0d6328f488ea76c571aa96fe3e4 : FAILURE in 1h 06m 44s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/4f477dc0935348218503a1c225c857bf : SUCCESS in 1h 14m 50s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/d9ad7322dbe34cd3a863592f6aeb4aaa : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f2dfbfac3cd1477b959157bbc554c16592e87246","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-19 10:56:29.000000000","message":"Patch Set 1:\n\nNot really needed right now: CS8 is used as an instance image, not for running OpenStack code. I\u0027d rather wait for RHEL 9 to go GA first.","accounts_in_message":[],"_revision_number":1},{"id":"cdcdb1cc95280a18e8f7bb349a6f389c9fdc133f","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-04-19 11:51:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d576db53145eb39ef17302d5b316c8d6e8f80bc2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-04-19 11:51:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b85954af85c8961283172876d574773a5aad7e29","revisions":{"b85954af85c8961283172876d574773a5aad7e29":{"kind":"REWORK","_number":1,"created":"2022-03-29 19:23:30.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/49/835749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/49/835749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/49/835749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/49/835749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/49/835749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/49/835749/1"}}},"commit":{"parents":[{"commit":"d06a066813f9c28e952325a069192668270519b2","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/d06a066813f9c28e952325a069192668270519b2"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2022-03-29 19:21:51.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2022-03-29 19:23:28.000000000","tz":-180},"subject":"Move from CS8 to CS9","message":"Move from CS8 to CS9\n\nSince we are now in Zed we should stop testing in CS8,\nmoving integration job to CS9\n\nChange-Id: I2c772b2d26bd7922dd9790a6edbeab76fcb1b62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b85954af85c8961283172876d574773a5aad7e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b85954af85c8961283172876d574773a5aad7e29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
