)]}'
{"id":"openstack%2Fmetalsmith~841911","triplet_id":"openstack%2Fmetalsmith~master~I6cad4556d770a733245d12f9a211e8ff7a3d884b","project":"openstack/metalsmith","branch":"master","topic":"bios","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-05-16 18:29:13.000000000","reason":"Harald Jensås replied on the change"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-05-17 21:58:11.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I6cad4556d770a733245d12f9a211e8ff7a3d884b","subject":"Add a CentOS job with legacy local boot","status":"MERGED","created":"2022-05-16 15:08:19.000000000","updated":"2022-05-17 22:00:19.000000000","submitted":"2022-05-17 21:58:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841911-bios","meta_rev_id":"8de1ffaca56d562491462e921e3c45c92218665e","_number":841911,"virtual_id_number":841911,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-17 21:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2022-05-16 18:29:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2022-05-17 20:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-17 14:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2022-05-17 20:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 15:08:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-16 15:23:39.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2022-05-17 14:54:33.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2022-05-17 20:28:02.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"6e67fee57ea1f2bcb8833cc5d0a86dac29ef6aaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-05-16 15:08:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e92cf7b74235aaa7900862a8a579caf534fd9b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 15:08:54.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/metalsmith on branch master.  The error was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  job:\n    name: metalsmith-integration-glance-centos8-legacy\n    parent: metalsmith-integration-glance-centos8-uefi\n  vars:\n  ...","accounts_in_message":[],"_revision_number":1},{"id":"d095b704dc995c117e9d7f7de0021ca5718231bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-05-16 15:18:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63c3d8601f03f19878740dcf76d9522a67d00482","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-16 15:21:58.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"1391d5a556abb7313dc8bb5d504fa13369900133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 16:11:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32837125a44f465f96a297d99756345b : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaf6f77e055a429e9a6e5ab33415ff68 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d6bc39ac46148c9a630f8169b0cc507 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83e497764e8a4595a17c29058733c145 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c0fdd54399a4bafa43d73f71ba755ab : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b496615847ba40f8b3eddcce85adc0de : SUCCESS in 8m 06s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/b7d5622fbff04b859e8689eb92c4b3e9 : FAILURE in 25m 17s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/100e8b6095b74fdeb9df8332c8fc0cc2 : FAILURE in 23m 15s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/78ae0687620348e8bfbf28f5f1bc850f : SUCCESS in 50m 42s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/5be7bfdb8731471c8266c3b1da136c43 : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":2},{"id":"bf646159911b590a51d7b2be97ea152dcbb883be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-05-16 16:17:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f8bdc43bb4c173059ddb3ee05ccc09c90672d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 18:11:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa03883ed90b4cb58438ae51944935b8 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7946189d7e8249ba9427518d2d8e7fc2 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d2ac1f82a824f9f9e3d7acd7cd36786 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b58276a28b4e42338f7ad7fa10e4a766 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc7d2edd7b7b48159c0818a08f7a3c21 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f579fa5373604d74a9d2dd62270715ec : SUCCESS in 9m 22s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/ebbb15d4f5c341b4b935eb4d30578233 : SUCCESS in 1h 46m 55s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/8b17c81c0c6b4ff1a1c1bf2f4b50eff9 : SUCCESS in 1h 49m 13s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/21d312325ee64c2e87eda34d0a8db3d2 : SUCCESS in 1h 04m 50s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/ade5cd3c717044408ccf98486573d93c : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":3},{"id":"8dcee6a00c89e24edd272af0e94968f356ba44d2","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-05-16 18:29:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"39946a6daa7a771412b2cd78d7c05774133b12bc","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-05-17 14:54:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"023dcbd6461dccaa1b1aae6077729eed8c4c7d33","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-05-17 20:28:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"00271edd69139e9733b9eb2f6d32afd6e1b741a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 20:28:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9f14289e2ffb7c8544a39dda4c7b6739ddbe16c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 21:58:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa7b7400a8c423c99134589feb6e860 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee488130a4fd41da9e0885900e20adf7 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e72a308713b4c5f98b3bbcd35c931d7 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eaf933f24ac449eb51f68ce3517e290 : SUCCESS in 8m 04s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/7c8895cf532f45789a26b5a4d66ba1b0 : SUCCESS in 1h 14m 24s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/eaf209c04ec543129e7937dc8eec59f3 : SUCCESS in 1h 28m 51s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/2b10ec0d5d9b485e80bfce7013431a1b : SUCCESS in 57m 31s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/dc317ab0715c4954ab1db8c67deac4d4 : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a1f69f9b93562c4cc03378b6d260bad08256cb49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 21:58:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8de1ffaca56d562491462e921e3c45c92218665e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 22:00:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e79cf4170044ed59a273d02cce94ccb : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a3e205fc059f456a677e42beeb512ef4f389e66b","revisions":{"49deba9c360e14ba1ca1d21748eb6667820f81e7":{"kind":"REWORK","_number":1,"created":"2022-05-16 15:08:19.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/11/841911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/11/841911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/1"}}},"commit":{"parents":[{"commit":"018a19c1e89d91d0468ca9a59ca4cf011a90a2ab","subject":"Add a workaround to fix the CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/018a19c1e89d91d0468ca9a59ca4cf011a90a2ab"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-05-16 15:06:08.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-05-16 15:07:39.000000000","tz":120},"subject":"Add a CentOS job with legacy local boot","message":"Add a CentOS job with legacy local boot\n\nWhen netboot is removed from Ironic, we need a job to cover local boot\nin legacy mode. The new job will do it.\n\nAlso drop \"local\" from the job name, it\u0027s the default.\n\nChange-Id: I6cad4556d770a733245d12f9a211e8ff7a3d884b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/49deba9c360e14ba1ca1d21748eb6667820f81e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/49deba9c360e14ba1ca1d21748eb6667820f81e7"}]},"branch":"refs/heads/master"},"16582168e8f953ac86965a501401f0be2030f0dd":{"kind":"REWORK","_number":2,"created":"2022-05-16 15:18:23.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/11/841911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/11/841911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/2"}}},"commit":{"parents":[{"commit":"018a19c1e89d91d0468ca9a59ca4cf011a90a2ab","subject":"Add a workaround to fix the CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/018a19c1e89d91d0468ca9a59ca4cf011a90a2ab"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-05-16 15:06:08.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-05-16 15:18:17.000000000","tz":120},"subject":"Add a CentOS job with legacy local boot","message":"Add a CentOS job with legacy local boot\n\nWhen netboot is removed from Ironic, we need a job to cover local boot\nin legacy mode. The new job will do it.\n\nAlso drop \"local\" from the job name, it\u0027s the default.\n\nChange-Id: I6cad4556d770a733245d12f9a211e8ff7a3d884b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/16582168e8f953ac86965a501401f0be2030f0dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/16582168e8f953ac86965a501401f0be2030f0dd"}]},"branch":"refs/heads/master"},"a3e205fc059f456a677e42beeb512ef4f389e66b":{"kind":"REWORK","_number":3,"created":"2022-05-16 16:17:05.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/11/841911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/11/841911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/11/841911/3"}}},"commit":{"parents":[{"commit":"018a19c1e89d91d0468ca9a59ca4cf011a90a2ab","subject":"Add a workaround to fix the CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/018a19c1e89d91d0468ca9a59ca4cf011a90a2ab"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-05-16 15:06:08.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-05-16 16:16:26.000000000","tz":120},"subject":"Add a CentOS job with legacy local boot","message":"Add a CentOS job with legacy local boot\n\nWhen netboot is removed from Ironic, we need a job to cover local boot\nin legacy mode. The new job will do it.\n\nAlso drop \"local\" from the job name, it\u0027s the default.\n\nChange-Id: I6cad4556d770a733245d12f9a211e8ff7a3d884b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a3e205fc059f456a677e42beeb512ef4f389e66b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a3e205fc059f456a677e42beeb512ef4f389e66b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
