)]}'
{"id":"openstack%2Fmetalsmith~849524","triplet_id":"openstack%2Fmetalsmith~stable%2Fxena~I02c1d7cc9f2f91e76d1b3edbc340df7ea83d2902","project":"openstack/metalsmith","branch":"stable/xena","topic":"add-linters-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2022-07-13 03:19:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02c1d7cc9f2f91e76d1b3edbc340df7ea83d2902","subject":"Add linters job","status":"MERGED","created":"2022-07-12 21:56:09.000000000","updated":"2022-07-13 03:23:21.000000000","submitted":"2022-07-13 03:19:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849524-add-linters-stable/yoga-stable/xena","cherry_pick_of_change":849523,"cherry_pick_of_patch_set":1,"meta_rev_id":"49afde00c072f8f8ee4cd4d44f9d44e5a61bb60d","_number":849524,"virtual_id_number":849524,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 03:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":2,"date":"2022-07-13 01:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":1,"date":"2022-07-13 01:16:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 21:56:09.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2022-07-12 23:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 01:16:11.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"3eb590f7747e6874d4d42c04a51af8faebe2e4e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-07-12 21:56:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"82f68c747ec6b366da835fb11c06a0f3fb332524","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 23:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/453187568fbd4f148745724b91c90998\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa499d857e944201a7ea7a358f9a4733 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/252538d60da0468cb0997f885168504e : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a38233eaeae34028b6124234643d8de8 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/851ec0f65d1446f0b15fce0f8ffeb868 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d3f3f7a87b948c09e7075bf904725a9 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c48d507de3415cad026ff1bd4f1cf6 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a1a8170e7e649b2a35ea6d5ebcba17a : SUCCESS in 5m 56s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/6ff6a0fe1dc5474583b5a049fdba0ebf : SUCCESS in 1h 30m 14s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/28a45a1fc0af470185d04e0d687d4cb4 : SUCCESS in 46m 12s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/9dad7965c71e48f189fd98d4141550e2 : SUCCESS in 54m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1df3f67ff017484699ec77dea5f43983 : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"df9170ba62183f4216bf50cdfcb6af461371a841","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-07-13 01:16:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e021400afa67205e8d8743e831bd8515d59921d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 01:16:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66d725c0ae6a7a4693063e3090e784cc55823cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 03:19:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e88354f73099479098040c65253c2a8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e43696791d2841ccbcb2f8c304270cfd : SUCCESS in 2m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3018aa178a6f4e62a873d8cedbdd4abd : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62eb50fb98834839a885d201360ddd7c : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6faea74a0455451d8c62f76afa03f807 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb1e4977070a44de9f771f54445694e0 : SUCCESS in 4m 04s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/e85f3c3bbb594e79b39a51f9759df73e : SUCCESS in 1h 52m 11s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/fbf3fabd8302439bba2027104769565a : SUCCESS in 1h 06m 25s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/f946098db5be453dafa4920e3eed4e8a : SUCCESS in 1h 06m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a84ecf4a3ee941b083adefc70976f60b : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a731b2c9d0d46771069ca3eba56a86766ad68cc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 03:19:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"49afde00c072f8f8ee4cd4d44f9d44e5a61bb60d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 03:23:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2775f8e4c248c4b5af9f14f4b65721\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528be4bcffe8420bb18c9efae579c7bc : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06c41fd53f2f4569aeb448b9dc12add7 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c6edd1769a13a9a1a07d67b99092ab41ae54a08","revisions":{"5c6edd1769a13a9a1a07d67b99092ab41ae54a08":{"kind":"REWORK","_number":1,"created":"2022-07-12 21:56:09.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/24/849524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/24/849524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/24/849524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/24/849524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/24/849524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/24/849524/1"}}},"commit":{"parents":[{"commit":"205a247d6b3cd93a1e20431b21a8e78db8aa07ba","subject":"metalsmith_instances module, unprovision by hostname","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/205a247d6b3cd93a1e20431b21a8e78db8aa07ba"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-07-12 07:59:49.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-07-12 21:56:09.000000000","tz":0},"subject":"Add linters job","message":"Add linters job\n\nFix lint errors to unblock CI\n\nChange-Id: I02c1d7cc9f2f91e76d1b3edbc340df7ea83d2902\n(cherry picked from commit f234b87bbbc3e8ed76247a453faf7421c4d9cfc1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/5c6edd1769a13a9a1a07d67b99092ab41ae54a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/5c6edd1769a13a9a1a07d67b99092ab41ae54a08"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}]}],"submit_requirements":[]}
