)]}'
{"id":"openstack%2Fmetalsmith~854239","triplet_id":"openstack%2Fmetalsmith~master~I5e65f9949ee3b78430309017800b88c1ba21dd02","project":"openstack/metalsmith","branch":"master","topic":"no-netboot","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-24 16:00:03.000000000","reason":"Change was submitted"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2022-08-24 00:05:43.000000000","reason":"\u003cGERRIT_ACCOUNT_15519\u003e replied on the change","reason_account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}},"hashtags":["ironic-week-prio"],"change_id":"I5e65f9949ee3b78430309017800b88c1ba21dd02","subject":"Deprecate instance network boot","status":"MERGED","created":"2022-08-23 15:59:03.000000000","updated":"2022-08-24 16:02:39.000000000","submitted":"2022-08-24 16:00:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"854239-no-netboot","meta_rev_id":"6561a7995eb6fc480f08eb6255890a0bc6a752a3","_number":854239,"virtual_id_number":854239,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 16:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2022-08-24 13:44:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2022-08-24 13:46:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-24 13:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2022-08-24 13:46:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 16:20:02.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2022-08-23 18:12:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 21:21:17.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2022-08-24 09:57:23.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-08-24 13:44:16.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"}],"messages":[{"id":"9fb937068fec5ab5ab10301840f40585bdbff382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-23 15:59:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df812b135ec573330223b4bfda5e542f172b1c08","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-23 16:20:10.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"320c88da69eb2c2d23207ae7cb3dd97ed8a273f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 18:12:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b0a3adbba6b4b6d8291c2e7caf8ad44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6242f6713b4f2c93848c977d3e901d : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47525672d5ad4ff4b90f0921c159c328 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c44cd788f174008ad7a849945981123 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7a5bf8bc75e4c9696dea0c382464bd4 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efa38f7b3b004643aaf75a9dee70c7c9 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6f72348d254a47ba684d000fd09576 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be604a76d5ed4859a0d0a49f1502d2d8 : SUCCESS in 4m 14s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/5eca7d6e02b141ba8f2225c4ed21c8af : SUCCESS in 2h 11m 03s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/8964ae460a3d47fa9a07764f2642743b : SUCCESS in 1h 47m 05s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/be50d4918a5844d1a8247dd45889b31b : FAILURE in 41m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e5252171977d46ecb63184b20b653f68 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"17f71dadd4274ddd1cc9a2cf0c66bcf3b4cb57ec","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-08-23 21:21:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"53bc6cef43759a6d0ea944d0cb06fab6b65bc5e9","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-24 00:05:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1e92ad25d625fe69c8bbeb8f918fd2f7621e739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-24 07:41:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ada387ef24e9f281b0662f1e9bb5a1ed06ea97fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-24 08:53:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0cb1b21d6efbfec7a3344064a9adca6f70be24d0","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-24 09:57:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"978c76610d7c0f65b9204d4bf6a3d24a62b3155f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 10:45:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4812570484ba4fffb9a62f4fb157d236\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7167c424b5734f3a98756cc0893c78ce : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74af94c40d6c4cf78c5f9d99ac69b136 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08db025d2313450e907017dec5a72748 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17bcd1abc0654fe492d9c93cecf2d8d6 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c1b1d79ce75427f8394d72fe4cedc78 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef150cf7b2bf4b8ba9705354028f471c : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a7b90ac42fa47c7a9e2b24530689364 : SUCCESS in 4m 17s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/516efd7d041f4108ad1917b14024b265 : SUCCESS in 1h 35m 51s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/7be98a3c4b81443d8b9919164ca22a8f : SUCCESS in 1h 33m 21s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/ce79a8d2c5904e25a608d73758797509 : FAILURE in 50m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/addeccfbaaea4889b6ffc46e0529e2b8 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":3},{"id":"4eefc2919914d7bc1353afcf4ac580b850e2acf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-24 11:19:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"22bd49872911037f3200502dfdebbbff16f422b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 13:17:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3feb68f6aba4ee0a72e84305d42c32f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6441df830cf1437799ea21fdfbad1783 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2937540327bf4d7688eab977c3206203 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d73ecf6339c742f08245b281e9c916d4 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac6211d000514e89a9e11ed59b929547 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5743dd852d82481ebcc40780c9a65466 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b7a91c2a384a03ac8f8f8c6a5c5e3e : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be4dee964a445d5a2f23e51b97ff208 : SUCCESS in 4m 03s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/c0efa08263d046b49159d30db0fe6430 : SUCCESS in 1h 53m 04s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/e1172ce1d5934effbe03f5bdb448ef38 : SUCCESS in 1h 43m 55s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/e99f9098c2da4cd285350e97ff4e532f : SUCCESS in 1h 09m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e44faf5777cf4485aab7176ebcab8f67 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":4},{"id":"520bcb4ded281b5900d6fc177e7c8a95a19bab9b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-24 13:37:20.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70460c2bec5d97e84fe64e24bd92687d17647f97","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-08-24 13:44:16.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3cdd28de7537f7693be355460103877cf51f78b2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-24 13:46:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"541208ce5063b142f7f740c903de10302ec1ce53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 13:47:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"702c0cb508365b94badd41f83c9814dd7d8cd100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:00:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4af21c9b78614e219057c37063ba6355\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74a5e3471ae48e1aa268ba65ad4a001 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f15e5e2fb02e41339e918ced97379d07 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85307c7cc1074d9b839c67bc1a29f0f3 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cd9580d07542d9845667ef75127950 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be82dc80f93e4aa0a0e7619a0fe6d640 : SUCCESS in 4m 03s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/d2d56c89f87b45a895d22131ed34d563 : SUCCESS in 1h 55m 46s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/8257e3e3c85f45ed8037e8965e8775ee : SUCCESS in 2h 05m 30s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/69b0d9c693c044d79ba97f54b0a66527 : SUCCESS in 1h 10m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/82a65413deb741f3bc1b57c880130c15 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":4},{"id":"e1a20c7ca38a9b42c234b18a7cba46ff90d81f28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:00:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6561a7995eb6fc480f08eb6255890a0bc6a752a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:02:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/027db6ce76dc464588483f918160211f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f11e5159cf4c8c8072cf02c9d11f54 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e659747ba60470e8cd05357a373454f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"77d802f3d6709c98a03813cd55651b4f70fbb805","revisions":{"1513611e230edbcf024e2f8df351decc521676d7":{"kind":"REWORK","_number":1,"created":"2022-08-23 15:59:03.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/39/854239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/39/854239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/1"}}},"commit":{"parents":[{"commit":"6c5e68a4433831e988845d5fca1e6880d169b6b7","subject":"Do a precise node lookup for unprovision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/6c5e68a4433831e988845d5fca1e6880d169b6b7"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-23 15:55:24.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-23 15:58:17.000000000","tz":120},"subject":"Deprecate instance network boot","message":"Deprecate instance network boot\n\nRemove or update the corresponding CI jobs. Use the new custom Cirros\nimage in the HTTP job to avoid relying on CentOS.\n\nTemporary increase complexity requirements while we have a deprecation\nwarning in the provisioner.\n\nChange-Id: I5e65f9949ee3b78430309017800b88c1ba21dd02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/1513611e230edbcf024e2f8df351decc521676d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/1513611e230edbcf024e2f8df351decc521676d7"}]},"branch":"refs/heads/master"},"0a98e258cf027eac568844ebd5c34e7f69a75bf1":{"kind":"REWORK","_number":2,"created":"2022-08-24 07:41:41.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/39/854239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/39/854239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/2"}}},"commit":{"parents":[{"commit":"6c5e68a4433831e988845d5fca1e6880d169b6b7","subject":"Do a precise node lookup for unprovision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/6c5e68a4433831e988845d5fca1e6880d169b6b7"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-23 15:55:24.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-24 07:41:35.000000000","tz":120},"subject":"Deprecate instance network boot","message":"Deprecate instance network boot\n\nRemove or update the corresponding CI jobs. Use the new custom Cirros\nimage in the HTTP job to avoid relying on CentOS.\n\nTemporary increase complexity requirements while we have a deprecation\nwarning in the provisioner.\n\nChange-Id: I5e65f9949ee3b78430309017800b88c1ba21dd02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0a98e258cf027eac568844ebd5c34e7f69a75bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0a98e258cf027eac568844ebd5c34e7f69a75bf1"}]},"branch":"refs/heads/master"},"780b54a92bc52d04c170c8095e53e848d88ff29e":{"kind":"REWORK","_number":3,"created":"2022-08-24 08:53:57.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/39/854239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/39/854239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/3"}}},"commit":{"parents":[{"commit":"6c5e68a4433831e988845d5fca1e6880d169b6b7","subject":"Do a precise node lookup for unprovision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/6c5e68a4433831e988845d5fca1e6880d169b6b7"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-23 15:55:24.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-24 08:53:41.000000000","tz":120},"subject":"Deprecate instance network boot","message":"Deprecate instance network boot\n\nRemove or update the corresponding CI jobs. Use the new custom Cirros\nimage in the HTTP job to avoid relying on CentOS.\n\nTemporary increase complexity requirements while we have a deprecation\nwarning in the provisioner.\n\nChange-Id: I5e65f9949ee3b78430309017800b88c1ba21dd02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/780b54a92bc52d04c170c8095e53e848d88ff29e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/780b54a92bc52d04c170c8095e53e848d88ff29e"}]},"branch":"refs/heads/master"},"77d802f3d6709c98a03813cd55651b4f70fbb805":{"kind":"REWORK","_number":4,"created":"2022-08-24 11:19:58.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/39/854239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/39/854239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/39/854239/4"}}},"commit":{"parents":[{"commit":"6c5e68a4433831e988845d5fca1e6880d169b6b7","subject":"Do a precise node lookup for unprovision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/6c5e68a4433831e988845d5fca1e6880d169b6b7"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-23 15:55:24.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-24 11:19:52.000000000","tz":120},"subject":"Deprecate instance network boot","message":"Deprecate instance network boot\n\nRemove or update the corresponding CI jobs. Use the new custom Cirros\nimage in the HTTP job to avoid relying on CentOS.\n\nTemporary increase complexity requirements while we have a deprecation\nwarning in the provisioner.\n\nChange-Id: I5e65f9949ee3b78430309017800b88c1ba21dd02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/77d802f3d6709c98a03813cd55651b4f70fbb805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/77d802f3d6709c98a03813cd55651b4f70fbb805"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
