)]}'
{"id":"openstack%2Fmetalsmith~860854","triplet_id":"openstack%2Fmetalsmith~master~I3c814b393a0b62666943dbf2ecf3e3481a80e90a","project":"openstack/metalsmith","branch":"master","topic":"test-ci","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-10-13 07:15:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3c814b393a0b62666943dbf2ecf3e3481a80e90a","subject":"[DNM] TEST CI","status":"ABANDONED","created":"2022-10-10 15:55:52.000000000","updated":"2022-10-13 07:15:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"83d926264920006fa65f438eeabce7a576ea3708","_number":860854,"virtual_id_number":860854,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-10 16:46:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-10 16:46:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0a4834fb0c5363771321e93ca2d19bc360cabfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-10 15:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"210a6eec05a19bb4e3a97b9e72913c365fb8e988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 16:46:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce7bac061ff444a088eed9516110ced9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/28f82a9fd7b7453387635bddb453e9da : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd54adc424b248d7a0ec9be7e8ac5f71 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2162e613ded448678cfa174057c7b2d1 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84a92e45d7ea443caea20e6df1d73f49 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7546678a9d374342b426ca00c83cd6ed : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24ccece9d78461eaed0b475f93937dd : SUCCESS in 6m 19s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/ce4f82986b3c461f97a7a448f234407e : FAILURE in 45m 30s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/fc83f1c44b634301a9144d8e20ced90b : FAILURE in 44m 32s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/c1df65a1b71742c1be907cab0fa52d60 : FAILURE in 39m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3f2448672ea04b3d80f5d76581449783 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"83d926264920006fa65f438eeabce7a576ea3708","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-13 07:15:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c631f360588694b738d9db27fdf2f4684f54de57","revisions":{"c631f360588694b738d9db27fdf2f4684f54de57":{"kind":"REWORK","_number":1,"created":"2022-10-10 15:55:52.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/860854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/54/860854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/54/860854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/54/860854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/54/860854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/54/860854/1"}}},"commit":{"parents":[{"commit":"cc8fe24d40e64e874bc3d4308cd76e50379729d3","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cc8fe24d40e64e874bc3d4308cd76e50379729d3"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-10-10 15:55:41.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-10-10 15:55:41.000000000","tz":120},"subject":"[DNM] TEST CI","message":"[DNM] TEST CI\n\nChange-Id: I3c814b393a0b62666943dbf2ecf3e3481a80e90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/c631f360588694b738d9db27fdf2f4684f54de57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/c631f360588694b738d9db27fdf2f4684f54de57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
