)]}'
{"id":"openstack%2Fmetalsmith~860885","triplet_id":"openstack%2Fmetalsmith~master~Ie026b1b9b4331fdd74f2e06784e8e017049603ce","project":"openstack/metalsmith","branch":"master","topic":"openstacksdk","attention_set":{},"removed_from_attention_set":{"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2022-10-11 19:21:02.000000000","reason":"\u003cGERRIT_ACCOUNT_4571\u003e replied on the change","reason_account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}},"hashtags":[],"change_id":"Ie026b1b9b4331fdd74f2e06784e8e017049603ce","subject":"Limit openstacksdk install to \u003c0.99","status":"ABANDONED","created":"2022-10-11 00:10:33.000000000","updated":"2022-10-11 19:21:11.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dac8ae13c142e323b37d20667dc7c6849193cf2d","_number":860885,"virtual_id_number":860885,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-11 01:24:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 01:24:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8a863ab811aefbb12b6fa6422fa1a12f1d38737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-10-11 00:10:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3db27ef427ce940d2ae8f8e2f8d2c8161973a32d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 01:24:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e7430c132034fe39cb4f278f9aa7729\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bff68aed059d460e8db1ff96afb146b7 : FAILURE in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f2ab3f9fc84e4e860e0aa79b30e862 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f5cc071e9e34ec9ab09b74872cfddc0 : FAILURE in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/709dee6bf5e74065b4736fd26b4ac8fe : FAILURE in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41625a83243244d3a593d162dbfccb99 : FAILURE in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d436603c867e44769f98ffb85be83b65 : FAILURE in 3m 08s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/8935d2d5ff86409b959cae2bc34790aa : SUCCESS in 1h 04m 54s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/2c2dade762e644f0b27cac3d76dde6e2 : SUCCESS in 1h 06m 00s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/5f520951400f40078323a1b2d2aae02f : SUCCESS in 50m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d0e73f052b71479ca518a636fd5aae7a : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"180f67f270c6c66e639af4510769f69ff12e3104","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-10-11 19:21:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dac8ae13c142e323b37d20667dc7c6849193cf2d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-10-11 19:21:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"400d888b37ef304948d79f1303b1884a56e8a863","revisions":{"400d888b37ef304948d79f1303b1884a56e8a863":{"kind":"REWORK","_number":1,"created":"2022-10-11 00:10:33.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/85/860885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/85/860885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/85/860885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/85/860885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/85/860885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/85/860885/1"}}},"commit":{"parents":[{"commit":"523cc1d49a77c8ea6c1136d970de0f6bce2ac55a","subject":"CI: Force ansible to version 5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/523cc1d49a77c8ea6c1136d970de0f6bce2ac55a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-11 00:00:15.000000000","tz":780},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-11 00:10:10.000000000","tz":780},"subject":"Limit openstacksdk install to \u003c0.99","message":"Limit openstacksdk install to \u003c0.99\n\nopenstacksdk version 0.99 onwards will become the stable 1.0.0 and\nincludes breaking changes.\n\nThe Ansible OpenStack collection now enforces what version of\nopenstacksdk it supports, see [1]. As long as we\u0027re consuming Ansible\nOpenStack collection stable 1.x we need to avoid installing\nopenstacksdk\u003e\u003d0.99.\n\nLikewise when we switch to Ansible OpenStack collection 2.x, this\nopenstacksdk version limit will need to be flipped.\n\nOur integration tests do not install their own ansible, so until that\nchanges, these tests are coupled with the ansible installed on the\nzuul image.\n\n[1] https://hackmd.io/szgyWa5qSUOWw3JJBXLmOQ?view\n\nChange-Id: Ie026b1b9b4331fdd74f2e06784e8e017049603ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/400d888b37ef304948d79f1303b1884a56e8a863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/400d888b37ef304948d79f1303b1884a56e8a863"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
