)]}'
{"id":"openstack%2Fmetalsmith~860886","triplet_id":"openstack%2Fmetalsmith~master~I4576ec5a623d053a4aa5505e27996c081273d8dd","project":"openstack/metalsmith","branch":"master","topic":"openstacksdk","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-10-11 18:30:21.000000000","reason":"Change was submitted"},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2022-10-11 18:30:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4576ec5a623d053a4aa5505e27996c081273d8dd","subject":"Revert \"CI: Force ansible to version 5\"","status":"MERGED","created":"2022-10-11 00:10:33.000000000","updated":"2022-10-11 18:32:08.000000000","submitted":"2022-10-11 18:30:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860886-openstacksdk","meta_rev_id":"e7cb0b735c81c37fc7dee5cf3bc13872fbf56da2","_number":860886,"virtual_id_number":860886,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-11 18:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2022-10-11 17:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2022-10-11 15:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2022-10-11 17:10:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 01:36:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-11 15:27:21.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-10-11 17:10:08.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"83a559d79d0db98a46c7acd847f1e2bfc4e2946a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-10-11 00:10:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3f9024cc53b9e43d214273b1137852766885cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 01:36:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f46243100d3d4d5888b2c363ccbae6ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32dab6b9ff1d4523987be9938de9282f : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0faf9a72653d474fbfbd1478786d0718 : FAILURE in 2m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28be3923fa204bed8f1ec447156b0c90 : FAILURE in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9bc76d979b842c3b93ddf360be3416b : FAILURE in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7c7bb1d4244bac8a027e60008aa3ad : FAILURE in 4m 27s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/89787429a1c54ab8a0a84f6605a6b6b1 : SUCCESS in 1h 24m 58s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/81b9f7142c8a497fa1aba06cc3d6adce : SUCCESS in 1h 13m 04s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/9267f2dff65f44c6a2887614941e979b : SUCCESS in 58m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/53afb68e05ec493eb577fd28045e9722 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"05fa1130968b575a3f66bf782df05c5ad4a02d1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-11 12:43:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3d93b3c775ef617dac9752faec7e3d51c273f1dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-11 12:46:17.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7ebfeb1063fc64b16310d336882cc36602f962ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 14:24:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9204cb1d73b549d1bc071d9d98ec639d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a9814c39784bdca5d90e0dbfed55cc : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dffc331aa6f4989b91c4f80c273e30e : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa12862db6ed4f73ae29e1d33ce00f11 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f16caaff6e64efa97e2c2a0560304a3 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e26feb059c5427d871bed1373e22479 : SUCCESS in 6m 29s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/e9f715c19e574213beebaf6d3edc52ea : SUCCESS in 1h 33m 21s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/d24b3f80ba0b4b07a8c1ceaa9ac1d4cf : SUCCESS in 1h 29m 32s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/3be9e960f3f34eb3835c1502f4b02b02 : SUCCESS in 1h 13m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/05d4f39678a34e51b3204527dec5eafc : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"20a56a2a10612bd372d4f3e1fc73b4bc489816fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-11 14:33:15.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"86660f66bc50370a0d905b96cb6b9419d1f104e5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-11 15:27:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2845f3dd9b17a06b76e329f999964337c1bf9648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 16:02:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0165eb19223434bafdb7af863ddd879\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe33a428d334b458681aff28afa09e2 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8ddbec750614e1ba4dccfe09d89e5af : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1820964cba1468db3fb91f2d0870710 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2806c50908f44b8b3943efdb8f65ed4 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/437f58aba0814eb39f3ff9cbf33c9a8a : SUCCESS in 6m 25s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/bd9ec42f01e64d20b176a3af48990717 : SUCCESS in 1h 17m 28s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/0a8435de85944a85953972721a7082a7 : SUCCESS in 1h 09m 03s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/0b0cb379d0c94d11a93e282a470d0823 : SUCCESS in 1h 23m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c779518382b44c8895b5dd576041b96f : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"619b5cda5860bf884fba92775e7f6714c1a83833","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-10-11 17:10:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"aa9d0f8fe5b1d28cceb9010902e9d767c38dd05e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:10:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ee7ecc3c5ebc7efbfa3cdba5a48cb4b052573f6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 18:30:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc9c7b75bafe40acaf4e86914929d60d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c9a175d6eed43d68f3b0608167b4ef6 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9e02879df5e4813a43841c6b057c1c4 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a3018a6146144198f7fc9973408ecc6 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff65304c52c4e8c8232c07c0cbe3e1d : SUCCESS in 7m 04s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/886c1a63ebbd4e3383d35326955cd22e : SUCCESS in 1h 13m 21s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/a2c0172e23794529a41b1fad74c05c9e : SUCCESS in 1h 18m 55s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/10d101dd1e1f4b61bcdcba91e2ce5e40 : SUCCESS in 52m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fcac884f17964558be4abeb3dc04816c : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":4},{"id":"095f8e395f050534dd29449a9009c7a623d371c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 18:30:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e7cb0b735c81c37fc7dee5cf3bc13872fbf56da2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 18:32:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a482be812e4534939f8a5ea72812f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903d8ab4141c4c3a9e3b19d52a1b923c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f988377a6ae32a749b92d8717ca0ec4fc713c924","revisions":{"481cdb6cf454619c38750b7c9cc2103397080e13":{"kind":"REWORK","_number":1,"created":"2022-10-11 00:10:33.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/86/860886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/860886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/1"}}},"commit":{"parents":[{"commit":"400d888b37ef304948d79f1303b1884a56e8a863","subject":"Limit openstacksdk install to \u003c0.99","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/400d888b37ef304948d79f1303b1884a56e8a863"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-11 00:09:28.000000000","tz":780},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-11 00:10:11.000000000","tz":780},"subject":"WIP Revert \"CI: Force ansible to version 5\"","message":"WIP Revert \"CI: Force ansible to version 5\"\n\nThis reverts commit 523cc1d49a77c8ea6c1136d970de0f6bce2ac55a.\n\nChange-Id: I4576ec5a623d053a4aa5505e27996c081273d8dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/481cdb6cf454619c38750b7c9cc2103397080e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/481cdb6cf454619c38750b7c9cc2103397080e13"}]},"branch":"refs/heads/master"},"277b30d29575154f2e746c969ee0f855a45cdd04":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-11 12:43:54.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/86/860886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/860886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/2"}}},"commit":{"parents":[{"commit":"ee008f7a79f079df98ec4a8556b3a067f4eb348a","subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ee008f7a79f079df98ec4a8556b3a067f4eb348a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-11 00:09:28.000000000","tz":780},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-10-11 12:43:54.000000000","tz":0},"subject":"WIP Revert \"CI: Force ansible to version 5\"","message":"WIP Revert \"CI: Force ansible to version 5\"\n\nThis reverts commit 523cc1d49a77c8ea6c1136d970de0f6bce2ac55a.\n\nChange-Id: I4576ec5a623d053a4aa5505e27996c081273d8dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/277b30d29575154f2e746c969ee0f855a45cdd04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/277b30d29575154f2e746c969ee0f855a45cdd04"}]},"branch":"refs/heads/master","description":"Rebase"},"e5d79a396fa8273d9aaf478bfc65c38719b459fc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-11 12:46:17.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/86/860886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/860886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/3"}}},"commit":{"parents":[{"commit":"ee008f7a79f079df98ec4a8556b3a067f4eb348a","subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ee008f7a79f079df98ec4a8556b3a067f4eb348a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-11 00:09:28.000000000","tz":780},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:46:17.000000000","tz":0},"subject":"WIP Revert \"CI: Force ansible to version 5\"","message":"WIP Revert \"CI: Force ansible to version 5\"\n\nThis reverts commit 523cc1d49a77c8ea6c1136d970de0f6bce2ac55a.\n\nDepends-On: https://review.opendev.org/c/openstack/metalsmith/+/860943\n\nChange-Id: I4576ec5a623d053a4aa5505e27996c081273d8dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e5d79a396fa8273d9aaf478bfc65c38719b459fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e5d79a396fa8273d9aaf478bfc65c38719b459fc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f988377a6ae32a749b92d8717ca0ec4fc713c924":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-10-11 14:33:15.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/86/860886/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/860886/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/860886/4"}}},"commit":{"parents":[{"commit":"ee008f7a79f079df98ec4a8556b3a067f4eb348a","subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ee008f7a79f079df98ec4a8556b3a067f4eb348a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-11 00:09:28.000000000","tz":780},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-10-11 14:33:15.000000000","tz":0},"subject":"Revert \"CI: Force ansible to version 5\"","message":"Revert \"CI: Force ansible to version 5\"\n\nThis reverts commit 523cc1d49a77c8ea6c1136d970de0f6bce2ac55a.\n\nChange-Id: I4576ec5a623d053a4aa5505e27996c081273d8dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f988377a6ae32a749b92d8717ca0ec4fc713c924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f988377a6ae32a749b92d8717ca0ec4fc713c924"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[]}
