)]}'
{"id":"openstack%2Fmetalsmith~860943","triplet_id":"openstack%2Fmetalsmith~master~Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2","project":"openstack/metalsmith","branch":"master","topic":"sdk-workaround","attention_set":{},"removed_from_attention_set":{"32962":{"account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"last_update":"2022-10-11 17:04:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2","subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","status":"MERGED","created":"2022-10-11 12:40:37.000000000","updated":"2022-10-11 17:05:40.000000000","submitted":"2022-10-11 17:04:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860943-sdk-workaround","meta_rev_id":"36c2cfb6713a6d30d845ddfd6360dc0e2ec86fea","_number":860943,"virtual_id_number":860943,"owner":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-11 17:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2022-10-11 14:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2022-10-11 14:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2022-10-11 15:27:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 14:29:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-11 14:32:31.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-10-11 14:37:19.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"26d74a6a9c618137a25a2acaf534660a39946f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-11 12:40:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f30a289053f2e611c2058424512373b29826589b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-11 12:41:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a382138eaaef7dcd490734f354348ad6bd44fee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-11 12:42:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7a06ad006b8089fc79f00daa8a5e4e1370302caa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 14:29:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3db1d9b760a40f9856cfe590c87abd4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9613575171ab4d0caad2d2db19d0496c : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e63952f01f84e50ab03916a45fb2579 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00a1b476fb094553af50bf7bdedfb8be : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c75081b91680473c945c27b8815b0432 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad66af0b7a740659c17970dba6eb3d2 : SUCCESS in 6m 15s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/296d78e4d2b84020855266a29730ed4a : SUCCESS in 1h 41m 03s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/57328c1cd18d4a949dab30f181776a18 : SUCCESS in 1h 38m 48s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/481449379a414c0785e5c6f6a6b661df : SUCCESS in 1h 19m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9590a14b75584219b7c44986c8f5650d : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":3},{"id":"57eec7353d3143cf1ce15af6e30cd96f48afbd80","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-11 14:32:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4859fcf8e3067c764465c833294e376d942f2f84","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-10-11 14:37:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e4280d178810ec0d6e9f39a07f3fc45dfb9ebe4d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-11 15:27:09.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d3e3e1ab1abaccc5dea1910ae48362f8159e7e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 15:27:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"57632177b41db4c87e43641c7a9c00ce6e15bb37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:04:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc125d4c76ab497383d47a60320d0b3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaadb07807734a04bc2bf2ca49852af0 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d883bc8b304f4ccaa1c0bf7e2745ae10 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe5d9a26550748589c707bbc9093d8d4 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e763c0bbd33a4e6ca1199ac865319a02 : SUCCESS in 6m 08s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/2541d030576f4c12b468dd47608c9976 : SUCCESS in 1h 32m 49s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/216e841350714de9b6708165eb182b00 : SUCCESS in 1h 32m 06s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/fc324368a28b4e508d4dd7620b3450f5 : SUCCESS in 52m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b70a27141c14e47acbcf9ab3444b430 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d643e37f031307821cbb5b92b0df953fabf44b07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:04:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"36c2cfb6713a6d30d845ddfd6360dc0e2ec86fea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:05:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b271e9c10991402baa3d9a2f61763aa4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff84ef9fa0b4b90b820ee1114239b04 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee008f7a79f079df98ec4a8556b3a067f4eb348a","revisions":{"6c5350eb8d8b9896efce56923eb65c72d6e00c30":{"kind":"REWORK","_number":1,"created":"2022-10-11 12:40:37.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/43/860943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/43/860943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/1"}}},"commit":{"parents":[{"commit":"523cc1d49a77c8ea6c1136d970de0f6bce2ac55a","subject":"CI: Force ansible to version 5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/523cc1d49a77c8ea6c1136d970de0f6bce2ac55a"}]}],"author":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:35:50.000000000","tz":120},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:40:36.000000000","tz":120},"subject":"Allow to use Ansible OpenStack Collection 1.x.x with openstacksdk \u003e\u003d0.99.0","message":"Allow to use Ansible OpenStack Collection 1.x.x with openstacksdk \u003e\u003d0.99.0\n\nAnsible OpenStack collection 1.x.x is not compatible to openstacksdk\n\u003e\u003d0.99.0, only collection series 2.x.x will be. But the latter has not\nbeen released yet.\n\nThis incompatibility only affects modules in Ansible OpenStack Collection but\nnot the functions that metalsmith uses. Those functions ARE compatible with\nrecent openstacksdk releases and will most likely not be removed in collection\nrelease 2.0.0.\n\nThus, as a hacky workaround to allow using Ansible OpenStack collection 1.x.x\nwith openstacksdk \u003e\u003d0.99.0 this patch removes the MAXIMUM_SDK_VERSION.\n\nChange-Id: Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/6c5350eb8d8b9896efce56923eb65c72d6e00c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/6c5350eb8d8b9896efce56923eb65c72d6e00c30"}]},"branch":"refs/heads/master"},"e7221b5ea3026c5459d454d09f9a295ce80e3e9d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-11 12:41:54.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/43/860943/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/43/860943/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/2"}}},"commit":{"parents":[{"commit":"523cc1d49a77c8ea6c1136d970de0f6bce2ac55a","subject":"CI: Force ansible to version 5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/523cc1d49a77c8ea6c1136d970de0f6bce2ac55a"}]}],"author":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:35:50.000000000","tz":120},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:41:41.000000000","tz":120},"subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","message":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0\n\nAnsible OpenStack collection 1.x.x is not compatible to openstacksdk\n\u003e\u003d0.99.0, only collection series 2.x.x will be. But the latter has\nnot been released yet.\n\nThis incompatibility only affects modules in Ansible OpenStack\nCollection but not the functions that metalsmith uses. Those\nfunctions ARE compatible with recent openstacksdk releases and\nwill most likely not be removed in collection release 2.0.0.\n\nThus, as a hacky workaround to allow using Ansible OpenStack\ncollection 1.x.x with openstacksdk \u003e\u003d0.99.0 this patch removes\nthe MAXIMUM_SDK_VERSION.\n\nChange-Id: Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e7221b5ea3026c5459d454d09f9a295ce80e3e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e7221b5ea3026c5459d454d09f9a295ce80e3e9d"}]},"branch":"refs/heads/master"},"ee008f7a79f079df98ec4a8556b3a067f4eb348a":{"kind":"REWORK","_number":3,"created":"2022-10-11 12:42:42.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/43/860943/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/43/860943/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/43/860943/3"}}},"commit":{"parents":[{"commit":"523cc1d49a77c8ea6c1136d970de0f6bce2ac55a","subject":"CI: Force ansible to version 5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/523cc1d49a77c8ea6c1136d970de0f6bce2ac55a"}]}],"author":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:35:50.000000000","tz":120},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:42:34.000000000","tz":120},"subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","message":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0\n\nAnsible OpenStack collection 1.x.x is not compatible to openstacksdk\n\u003e\u003d0.99.0, only collection series 2.x.x will be. But the latter has\nnot been released yet.\n\nThis incompatibility only affects modules in Ansible OpenStack\nCollection but not the functions that metalsmith uses. Those\nfunctions ARE compatible with recent openstacksdk releases and\nwill most likely not be removed in collection release 2.0.0.\n\nThus, as a hacky workaround to allow using Ansible OpenStack\ncollection 1.x.x with openstacksdk \u003e\u003d0.99.0 this patch removes\nthe MAXIMUM_SDK_VERSION.\n\nChange-Id: Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ee008f7a79f079df98ec4a8556b3a067f4eb348a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/ee008f7a79f079df98ec4a8556b3a067f4eb348a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
