)]}'
{"id":"openstack%2Fmetalsmith~861055","triplet_id":"openstack%2Fmetalsmith~stable%2Fxena~Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2","project":"openstack/metalsmith","branch":"stable/xena","topic":"sdk-workaround-stable/xena","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-12-16 19:42:12.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2","subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","status":"MERGED","created":"2022-10-13 07:06:07.000000000","updated":"2022-12-16 19:43:13.000000000","submitted":"2022-12-16 19:42:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861055-sdk-workaround-stable/xena","cherry_pick_of_change":860943,"cherry_pick_of_patch_set":3,"meta_rev_id":"69391e0b1a19225bc3d8d8746b35f72b88d4252b","_number":861055,"virtual_id_number":861055,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-16 19:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2022-10-14 12:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2022-12-16 18:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2022-12-16 18:28:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}],"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-13 07:06:07.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"state":"CC"},{"updated":"2022-10-13 08:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 12:28:05.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2022-12-16 18:28:29.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"1b09b92a2efb2024cbbd89eb5c07fb3bd19fd5cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-13 07:06:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4913bcbec61a4e4e213960cc8491a56375a61a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 08:53:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9075934eccbe4cbca6ea87cf4af18c70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a896038a771e45fd8a2551097e852599 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b3e8aacc31542f58f0033d2d441e1c7 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f34c637ade24db69ddf8b4fe1f85464 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20592ff6a7604ee69b7556d6250af0ce : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e2788b7a3a9450e8192adf7629b50a0 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cde8bd1fbf24a1fa06ccd8f6c3a1a39 : SUCCESS in 6m 42s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/5ac19d0fad104b3b9573b4efb4c23e21 : SUCCESS in 1h 43m 10s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/96d50a19ac554a92b107f98c870f8b13 : SUCCESS in 50m 40s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/0d5be2020d414b43a5728b9b72561f96 : SUCCESS in 1h 11m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/32726291807e4c72a9b4f84b7a315d89 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"61ce214ea4f441ae142509107e926a23bf520b7a","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-10-14 12:28:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a5eb8cc6c6b9e42610ac685960c0f15edcd13cd1","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-16 11:07:40.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"fde9521688339cfaec89508b51ff9546968e3b44","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-12-16 18:28:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5d8cdc969a6b88dbd5f7cd8a0741b27a68126a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 18:29:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddf859d69fc1c4fe20e39a5107d9ca50b81691ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 19:42:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f16f441fb545b1806bed4954cb7d31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19da908dcc304c66bb6e9fbe89d3724a : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94455bc4122644508b15ac9c67386a55 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a218aa1c34dc4505b1f675929fe7f2df : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b189d7cc0b8f499b874d0ab09f1eb1b8 : SUCCESS in 6m 17s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/e0a84f57c8d34600b263d28bac547750 : SUCCESS in 1h 11m 51s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/92c9ced36f984cd6a1ee97661f2381ac : SUCCESS in 1h 07m 54s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/e6c4b91821c346dbaa95d6b5a3919398 : SUCCESS in 46m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/83b7d6a60ea14235ae6ed80651aa9a7b : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5d3759e56cf51f6f7fe7e01b2602e8eaf23ac574","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 19:42:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"69391e0b1a19225bc3d8d8746b35f72b88d4252b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 19:43:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deac20a63319470f9c8b7f5e3e412963\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55d301950f442569941c22c264d3fed : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c889c779a3ef2b46850867b21c8a42eac13aaa17","revisions":{"c889c779a3ef2b46850867b21c8a42eac13aaa17":{"kind":"REWORK","_number":1,"created":"2022-10-13 07:06:07.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/55/861055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/55/861055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/55/861055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/55/861055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/55/861055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/55/861055/1"}}},"commit":{"parents":[{"commit":"46c6c29d2a9eb5544a6d12a49aa40f4fb3ab8b58","subject":"Merge \"Do a precise node lookup for unprovision\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/46c6c29d2a9eb5544a6d12a49aa40f4fb3ab8b58"}]}],"author":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:35:50.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-10-13 07:06:07.000000000","tz":0},"subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","message":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0\n\nAnsible OpenStack collection 1.x.x is not compatible to openstacksdk\n\u003e\u003d0.99.0, only collection series 2.x.x will be. But the latter has\nnot been released yet.\n\nThis incompatibility only affects modules in Ansible OpenStack\nCollection but not the functions that metalsmith uses. Those\nfunctions ARE compatible with recent openstacksdk releases and\nwill most likely not be removed in collection release 2.0.0.\n\nThus, as a hacky workaround to allow using Ansible OpenStack\ncollection 1.x.x with openstacksdk \u003e\u003d0.99.0 this patch removes\nthe MAXIMUM_SDK_VERSION.\n\nChange-Id: Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2\n(cherry picked from commit ee008f7a79f079df98ec4a8556b3a067f4eb348a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/c889c779a3ef2b46850867b21c8a42eac13aaa17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/c889c779a3ef2b46850867b21c8a42eac13aaa17"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[]}
