)]}'
{"id":"openstack%2Fmetalsmith~861056","triplet_id":"openstack%2Fmetalsmith~stable%2Fwallaby~Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2","project":"openstack/metalsmith","branch":"stable/wallaby","topic":"sdk-workaround-stable/wallaby","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-12-16 19:43:12.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2","subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","status":"MERGED","created":"2022-10-13 07:06:21.000000000","updated":"2022-12-16 19:44:27.000000000","submitted":"2022-12-16 19:43:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861056-sdk-workaround-stable/wallaby","cherry_pick_of_change":860943,"cherry_pick_of_patch_set":3,"meta_rev_id":"3b3b4d4e61293cfe52a7cacf8f5fbe5099ffbc9a","_number":861056,"virtual_id_number":861056,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-16 19:43:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2022-10-14 12:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2022-12-16 18:28:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2022-12-16 18:28:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}],"CC":[{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-13 07:06:21.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"state":"CC"},{"updated":"2022-10-13 09:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 12:28:41.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2022-12-16 18:28:38.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"1abb6794b434844eb660126a70d2781e22f2d650","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-13 07:06:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cb38a6c7ac7f35fcb94e807d48a44cdb57986647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 09:15:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12fa64e73f9e410e891690d0c360cf74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9452dc894c489aa698bf7ec9e7cf18 : SUCCESS in 2m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90476a1a3a4a4fad8cca2d8463ccddc8 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/320d0d6c236842dbb8a4b27bece68a1d : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fb7bea6d312484797ed15b949b3e6b0 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40d756992998421382372e98ce3525f3 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d5c6c47e1946d697da88eba87a32ef : SUCCESS in 8m 29s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/602b0113c1b445a0a766c1571a9a274c : TIMED_OUT in 2h 02m 15s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/b6b16e707edf40e19edcf85f50f2f2a2 : SUCCESS in 1h 05m 06s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/8b8759875e73417bba51ee80f5bb54a1 : SUCCESS in 1h 13m 06s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/27e636615b9542c7a72fa17b81b6fc8d : SUCCESS in 1h 08m 21s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/767d9562f384479c90cb3307cf2aa4f3 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c4c608ab068f57b3c04029e56a969b9121f96d8b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-13 09:51:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0188a0f8744c8c7a4a06e30c289951c971af58da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 11:12:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8686abd83a374a7c833756d12777c8a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd173bae36248ffa436b3db39d8ba90 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42adeaefe5b94b458a39f8694f3333ac : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c540bb264d2f497cbf78fce8abc23509 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b89589ff67fa47b792e583d63c3513b3 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56ba883f53014c46af1dcebdc419e097 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecf8de22c9394babb3782e3152203b2c : SUCCESS in 8m 43s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/2e8bbdf516834aebb52d19836dc6bd16 : SUCCESS in 1h 16m 01s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/8fa7dfa9c1cb4a0ca164153d995437e7 : SUCCESS in 54m 03s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/894e65d9bd624f338d9df5971fad1a1a : SUCCESS in 50m 49s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/b939002f7805472fa1244ab65e246e75 : FAILURE in 38m 33s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b0df7cc35b37440a8052c895e440d655 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"605e7cc484fa51229ff17fe29db69ec7d5be8930","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-10-14 12:28:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba3f42a840acb2796dd8f3b210d4f9fb5fd00bbf","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-16 11:07:35.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"c41baca6da72c0c5c4f81f7fb0a5f1e225611297","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-12-16 18:28:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64d9a13ac85c199da95035d2d00994e024c853cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 18:29:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eaa4c82f808e7b23f5f5d2973d4e8e885ec40e6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 19:43:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dabb0aa38434986af1e806a1d09b6b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4fd2b5571748479a2188cc25f4c294 : SUCCESS in 2m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc19e630ba4d4c529280b64f779b7745 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2611d26daf5e4db7a08984ee04e5cf70 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e1111cbd9a4ecc9c877d7b35dee82f : SUCCESS in 10m 07s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/be7c896d0689447c90ca12171b7b7684 : SUCCESS in 1h 13m 07s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/7db29252387a40a287b2192c5c55861e : SUCCESS in 45m 23s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/dccece694f2342308e2e058063157409 : SUCCESS in 44m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2e260bfcb6ab496ba08539d356d30cb8 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cf5bd43c25a09388b39e28cbed5e906dfdb4ed62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 19:43:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3b3b4d4e61293cfe52a7cacf8f5fbe5099ffbc9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 19:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e25d3f0d0f840d7bf9fadd30557863a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45776059b62b4461b969750076636c93 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cffe7d5093151632a78fa22faf44fd363adec05e","revisions":{"cffe7d5093151632a78fa22faf44fd363adec05e":{"kind":"REWORK","_number":1,"created":"2022-10-13 07:06:21.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/56/861056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/56/861056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/56/861056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/56/861056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/56/861056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/56/861056/1"}}},"commit":{"parents":[{"commit":"5e7461e9d2d4156390c4809654094a26148fb054","subject":"Merge \"Do a precise node lookup for unprovision\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/5e7461e9d2d4156390c4809654094a26148fb054"}]}],"author":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-11 12:35:50.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-10-13 07:06:21.000000000","tz":0},"subject":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0","message":"Allow to use Ansible OpenStack Col. 1.x.x with openstacksdk \u003e\u003d0.99.0\n\nAnsible OpenStack collection 1.x.x is not compatible to openstacksdk\n\u003e\u003d0.99.0, only collection series 2.x.x will be. But the latter has\nnot been released yet.\n\nThis incompatibility only affects modules in Ansible OpenStack\nCollection but not the functions that metalsmith uses. Those\nfunctions ARE compatible with recent openstacksdk releases and\nwill most likely not be removed in collection release 2.0.0.\n\nThus, as a hacky workaround to allow using Ansible OpenStack\ncollection 1.x.x with openstacksdk \u003e\u003d0.99.0 this patch removes\nthe MAXIMUM_SDK_VERSION.\n\nChange-Id: Icbcb0f92fc1e4315e81fe9ce0c2f9ce6b39867c2\n(cherry picked from commit ee008f7a79f079df98ec4a8556b3a067f4eb348a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cffe7d5093151632a78fa22faf44fd363adec05e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cffe7d5093151632a78fa22faf44fd363adec05e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[]}
