)]}'
{"id":"openstack%2Fmetalsmith~867851","triplet_id":"openstack%2Fmetalsmith~stable%2Ftrain~I687568e85656d984ee69ac62e751c4fcbea7acaa","project":"openstack/metalsmith","branch":"stable/train","hashtags":["ironic-week-prio"],"change_id":"I687568e85656d984ee69ac62e751c4fcbea7acaa","subject":"Fixes for tox 4.0","status":"MERGED","created":"2022-12-15 18:24:26.000000000","updated":"2022-12-15 19:15:35.000000000","submitted":"2022-12-15 19:14:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867851","cherry_pick_of_change":867850,"cherry_pick_of_patch_set":1,"meta_rev_id":"a346ff04f6f41ea3e14046511b080fbdd1c50f78","_number":867851,"virtual_id_number":867851,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-15 19:14:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-15 19:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-15 19:00:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 18:38:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-15 19:00:35.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"864d9ef683125039c0d32f091dce3db25ad01b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-12-15 18:24:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"5ba3e13e98fe291f1e4868ba33103901c71e3391","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-12-15 18:24:39.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"49ab7ab531c447e733df178def4be44a2e65d7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 18:38:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/984f5282bc05429c8a53e46ea11e25b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a5de30dadaf463a95a8d12bbe15bf4e : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c315173c28d14239913b8fcf395b06e9 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4840f3014bff43328ec78b97e8b280d2 : SUCCESS in 2m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f953d3fd551a437896508f035c492965 : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a695441892fd4596b1f972b96feace79 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8fc3ae4dc844d088a6b8540e03c250f : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad155b1be64f4970bba4818305a1473f : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/154cba2a80aa4ef4a04a282a54e347f0 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"49ffc5af208cac2a20db9394053b703cf9a4dd62","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-12-15 19:00:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8de7966977d411aebeaa245cdfafb3622cc37493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:01:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81b821ac4dfabeac3dfef09d451c11fc8f56252e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:14:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d47b9295bde4e74bf26611f1b6b25f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2687952d0be44c0839c0523a4963c3f : SUCCESS in 2m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9cde378a6c84d8ebde7859f157312a5 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cb9b9d0caf847bdaaae97e36eeec4c5 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe3f9a6c924541538c2f1ba9ebfb5b21 : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/547cc3fa3e334b3f8e4579d5f619a00f : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac353ea3a56d4e42a482a971cde5de39 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d133d78c8d8480584c3f9300da25bfa : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c036cfe4dd3950446b1fbc77fe342839e075581a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:14:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a346ff04f6f41ea3e14046511b080fbdd1c50f78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:15:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d9dc6779db042a1a110dd9277bc3391\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e2b9835b264b84a5f15d1f5d26708f : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d348d0897ea4d19b3f35cc07ba7980d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f16854ffa15c101e9fe0c4663cb7e71145296081","revisions":{"f16854ffa15c101e9fe0c4663cb7e71145296081":{"kind":"REWORK","_number":1,"created":"2022-12-15 18:24:26.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/51/867851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/51/867851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/51/867851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/51/867851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/51/867851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/51/867851/1"}}},"commit":{"parents":[{"commit":"8a8033a9fb442ac46760ed7a9ba4ef90652515e2","subject":"Ensure tox.ini is ASCII","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/8a8033a9fb442ac46760ed7a9ba4ef90652515e2"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-12-15 17:42:55.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-12-15 18:24:26.000000000","tz":0},"subject":"Fixes for tox 4.0","message":"Fixes for tox 4.0\n\nBefore this, the config would not parse with tox \u003e\u003d4.0\n\nChange-Id: I687568e85656d984ee69ac62e751c4fcbea7acaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f16854ffa15c101e9fe0c4663cb7e71145296081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/f16854ffa15c101e9fe0c4663cb7e71145296081"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[]}
