)]}'
{"id":"openstack%2Fmetalsmith~876671","triplet_id":"openstack%2Fmetalsmith~stable%2Fyoga~Ic4cef61e338fb57bea5de4f96eb3584386cc5152","project":"openstack/metalsmith","branch":"stable/yoga","topic":"story/2010571-stable/zed-stable/yoga","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2023-03-09 18:51:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic4cef61e338fb57bea5de4f96eb3584386cc5152","subject":"list_instances - cache allocations","status":"MERGED","created":"2023-03-07 09:05:49.000000000","updated":"2023-03-09 18:51:59.000000000","submitted":"2023-03-09 18:51:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876671-story/2010571-stable/zed-stable/yoga","cherry_pick_of_change":873589,"cherry_pick_of_patch_set":1,"meta_rev_id":"02407b494d455a7b4eb4ed8263534b078f0bca1f","_number":876671,"virtual_id_number":876671,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 18:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-03-07 14:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2023-03-09 04:20:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2023-03-09 04:20:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-07 10:51:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-07 14:18:20.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-03-09 04:20:13.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"1240cabc08884c2e8499eb96e3bd2f6a47c11b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-03-07 09:05:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"340701984b7970e254238ed8a47df643b7897e99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 10:51:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23868b4395844836ae29642a7e559416\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b12d70e67ff946f2a118eb5e532a5c68 : SUCCESS in 2m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3b610be807347f2917d5a10eb5938f6 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92cf1440cbad4a46944f1371b01e442e : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b0909ae606140cebc9a886a49d86f98 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b102afb2e1f1487e8ddedfcc4fe71d41 : SUCCESS in 6m 19s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/000ba33737c0472994c1a8f770afe641 : SUCCESS in 1h 35m 36s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/0ba37a82c0ca4a58b95c5155b087a58e : SUCCESS in 1h 03m 15s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/ef6d5e48f35c4ba3853352d41cae8b04 : SUCCESS in 1h 01m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d257ed3ecb8415891be56a93b4f403b : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9d695763911ca75dc4cb92c2827f6a25a2fdecc8","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-03-07 14:18:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33386543ae8d98514c7026de2fe4a5382939aed3","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-03-09 04:20:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3b83e7c2cd6c3a313e6adaff63b33b83066b06c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 04:20:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"232ed03518fc1e4ba40479c43a1b23e9db394c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 07:21:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a132e0f2074453fb00d0a836d151a6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf33529ba6a4a4c804b56dada0bafc4 : SUCCESS in 2m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab9960900a20423893afb5886f65add0 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/439006dce6ba439c87e2594447ed8f32 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d446215641e043bc9c97cbbf72c33069 : SUCCESS in 5m 47s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/b4541d7ee9dd4f1d82500d0b774d8577 : SUCCESS in 1h 13m 28s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/5af65efe9cd0466aae3f7ac83f50518d : SUCCESS in 42m 30s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/3c5d97104c3043ca9390a09d87261496 : SUCCESS in 42m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9dc7422aed3d4f36b9874c1a57414391 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"40b03949637c39ac0b78ebe3fa9454aee84f1fef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 17:08:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ae5a998b40269b495ce1788170111bae3ce7a07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:50:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea2b54139150449783ddf3df93cd7bc9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac648d126cb948e387cb3aa8f264a413 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f03a07ee315a43a8a75e230fdb109508 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5d826daefe145069f9f63056f385a37 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10bdc8d1146a4dbf8afaedcd5aec83eb : SUCCESS in 10m 59s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/bc9d4aaccb484590a5376eff7daa4116 : SUCCESS in 1h 32m 25s\n- metalsmith-integration-glance-netboot-cirros https://zuul.opendev.org/t/openstack/build/bc6346ec5af94b06b13a5c6a10a09daf : SUCCESS in 51m 02s\n- metalsmith-integration-http-netboot-cirros https://zuul.opendev.org/t/openstack/build/fa0dd91a114e4298a36b9f4fe27aaee9 : SUCCESS in 52m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7b5eec68c7b9491abd0a84c9a6b80b9d : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"56c5a77e17b6b99421bcce73f5f6457e63f4fa37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:51:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"02407b494d455a7b4eb4ed8263534b078f0bca1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc50c661f1ac45e9a206f8f7e0c104ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ca7dc213794d6db68208aa4fe5339e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc9f88b7e70381eeb417f4540574aa85a9dc41fa","revisions":{"cc9f88b7e70381eeb417f4540574aa85a9dc41fa":{"kind":"REWORK","_number":1,"created":"2023-03-07 09:05:49.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/71/876671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/71/876671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/71/876671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/71/876671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/71/876671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/71/876671/1"}}},"commit":{"parents":[{"commit":"04aebc7c4b5b0d13af2e27bdf683df1b8bbef6d1","subject":"Use a network cache in Instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/04aebc7c4b5b0d13af2e27bdf683df1b8bbef6d1"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-02-08 00:34:10.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-03-07 09:05:49.000000000","tz":0},"subject":"list_instances - cache allocations","message":"list_instances - cache allocations\n\nWhen listing instances fetch all allocations to a cache,\nfetch each nodes allocation from the cache. This is\nsignificantly faster than calling get_allocation for each\nnode.\n\ntime metealsmith list # 250 nodes:\n  real 0m54.970s vs real 1m45.048s without \"cached\".\n\nStory: 2010571\nTask: 47318\nChange-Id: Ic4cef61e338fb57bea5de4f96eb3584386cc5152\n(cherry picked from commit 364fa239165702d5cedc134666046aa9de1a595b)\n(cherry picked from commit 392593dd7707f7d7b90c2a96c14f7eb984d1d0d4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cc9f88b7e70381eeb417f4540574aa85a9dc41fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cc9f88b7e70381eeb417f4540574aa85a9dc41fa"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
