)]}'
{"id":"openstack%2Fmetalsmith~887867","triplet_id":"openstack%2Fmetalsmith~stable%2Fwallaby~Ied2d16ec33fe71522c3461d3df6e70fbfdd976b2","project":"openstack/metalsmith","branch":"stable/wallaby","topic":"story/2009732","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2023-07-09 23:02:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied2d16ec33fe71522c3461d3df6e70fbfdd976b2","subject":"Allow both \u0027network\u0027 and \u0027subnet\u0027 in NIC","status":"MERGED","created":"2023-07-06 18:06:06.000000000","updated":"2023-07-09 23:03:11.000000000","submitted":"2023-07-09 23:02:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887867-story/2009732","meta_rev_id":"51efe0400cc48581e6d2cd8b85462c8b988c6b7e","_number":887867,"virtual_id_number":887867,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-09 23:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-07-06 18:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2023-07-09 21:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2023-07-09 21:56:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 18:07:38.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-07-06 19:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-09 21:56:54.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"0ef2a951977469caaabf05af8d7ea612dd343f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-07-06 18:06:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7db6aa46b1ed49dd08d9a2f7891308ce85f8e1b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-07-06 18:07:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"147e179cd28373adbad460aeca98b58b633952d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 19:15:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3de74ce7393420fb7e11a336f6e9439\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5fd1e0a02954c119bfaf5c5b575b1c0 : SUCCESS in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b7734b3b68f4f36a7a725cc001a330d : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75b447740be6474bbd1ec5b1ebce1e24 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04e616fcae784cb4b54f6531900b42ab : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9d1a7d10fa44a1292f178903f9d4ef8 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68051c56cb88430fa4f3147df54d89d2 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4d239363a094a9b8ae7c7850f08fc63 : SUCCESS in 3m 03s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/843a2ccd39e0439ebf65648dd112e7a9 : SUCCESS in 1h 08m 26s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/749f4579b5224715b2894d702d7e7fe2 : SUCCESS in 43m 45s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/a09c0c9c2b9c4398891d06bbff212ef7 : SUCCESS in 46m 54s\n- metalsmith-integration-glance-netboot-cirros-iscsi https://zuul.opendev.org/t/openstack/build/b9fedfcf76414aa2a32eb82fa986387c : SUCCESS in 44m 09s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2890b77b167544418533a8f7cbfba0f4 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d5a212ecaab3deec61f42a185addfb201d28f97c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-07-09 21:56:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9eac008c88093e73f2d558a14c1186011bbe6a78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 21:57:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b53974a7b95f198f920838f03108ba9be61d2b99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 23:02:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/548e71806c56417e8b279ff3f71dc70f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82302dd0e4b24b3aaa220742c8b81710 : SUCCESS in 2m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10fa8552d8ee4cc2a299ea34b4989cd4 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7319ac00d97645b5a2d7bf8040196971 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2758e703df8f49fda612becab57c30e9 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f252850ee8cb418b81c3adbaaf1b673f : SUCCESS in 4m 42s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/008526b78f2d4751afe75a4f7d181cd0 : SUCCESS in 1h 04m 24s\n- metalsmith-integration-glance-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/36fabcef97a54eb49790998be442552f : SUCCESS in 42m 48s\n- metalsmith-integration-http-netboot-cirros-direct https://zuul.opendev.org/t/openstack/build/f3d671a1310d49b4a4db467dd79c1e3f : SUCCESS in 41m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3adb33942c3a42d68f9c4ef90d6e11c4 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cd8e0bd5539991e641bc10474c0bc5d7d01ffda0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 23:02:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"51efe0400cc48581e6d2cd8b85462c8b988c6b7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 23:03:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/110cfcd5ad1f4a3a92f781cd9eb7c215\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9b6bcc098645f382c82a62d165d7dd : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebdf953d4be54eb3b46de7c7d376dfc4 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"554df57f34352f5e6e7daeca510f049d218874e4","revisions":{"554df57f34352f5e6e7daeca510f049d218874e4":{"kind":"REWORK","_number":1,"created":"2023-07-06 18:06:06.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/67/887867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/67/887867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/67/887867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/67/887867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/67/887867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/67/887867/1"}}},"commit":{"parents":[{"commit":"1628113fc4e4016e6620e0774495f04f34b4610b","subject":"Get ports by \u0027binding:host_id\u0027 query filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/1628113fc4e4016e6620e0774495f04f34b4610b"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-12-09 14:20:29.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-07-06 18:04:52.000000000","tz":120},"subject":"Allow both \u0027network\u0027 and \u0027subnet\u0027 in NIC","message":"Allow both \u0027network\u0027 and \u0027subnet\u0027 in NIC\n\nFixes and issue where a port cannot be created on\na specific subnet if there are multiple subnets\nwith the same name on different networks.\n\nAllows both \u0027network\u0027 and \u0027subnet\u0027 in NIC information,\nwhen looking up the subnet filter on the network_id when\nboth \u0027network\u0027 and \u0027subnet\u0027 is provided.\n\nConflicts:\n\tmetalsmith/test/test_nics.py\n\n  Unit test for nics was added after wallaby.\n\nStory: 2009732\nTask: 44152\nChange-Id: Ied2d16ec33fe71522c3461d3df6e70fbfdd976b2\n(cherry picked from commit 264836d59ac741424c3fad4d47e51073722c848f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/554df57f34352f5e6e7daeca510f049d218874e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/554df57f34352f5e6e7daeca510f049d218874e4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
