)]}'
{"id":"openstack%2Fmetalsmith~895703","triplet_id":"openstack%2Fmetalsmith~stable%2Fzed~I67a3397f53e03270ac3ce540042f188991cd2bac","project":"openstack/metalsmith","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-09-18 19:37:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67a3397f53e03270ac3ce540042f188991cd2bac","subject":"stable-only: Constrain the upper Ansible version","status":"MERGED","created":"2023-09-18 16:51:16.000000000","updated":"2023-09-18 19:38:44.000000000","submitted":"2023-09-18 19:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895703","meta_rev_id":"47a0087fedf4098fbea317baf8b962d7f87f1727","_number":895703,"virtual_id_number":895703,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-18 19:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2023-09-18 19:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2023-09-18 19:20:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 17:04:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-18 19:19:48.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"df9f8ba924f8feedb8ffd26b12da64e2d3d0e7e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-18 16:51:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"333d780a4f7f7cdf885383e35ec1a3fed41fddd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 17:04:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e64f936431b74ccf8b2db0d1674f4e51\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/81afbda54151442c8c4e7a77979c7a2b : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0180d5de19b54e9e88f59036d742c87b : SUCCESS in 2m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dbef0f6e12f4179b81aca995c4150a9 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73db4cfeac6f4b59b7913073e8aaeaa6 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12a4409821f241b7afce3b464605c436 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/003ead0cd7d84d5281cb2ce1482dbad9 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c547215e2524bf99ee1725348f6b0cf : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4ff6c1d495e4449b87d398f046dd230 : SUCCESS in 3m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/feffb360389d451b9ec0bac9c0f224b2 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f682781f1d19361a1ce04689ab2a66500d30feed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-18 19:18:04.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cf1699a998413d842817caeea84eb35b27fd0602","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-18 19:18:47.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c91968c615c1491f5e6bcf857b9a7e8025769d0e","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-18 19:19:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e4d140c94c82bf5ade45762a76d4c2abbfd75190","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-18 19:20:00.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"097a3a12136cdb1b189f4473d9cde15103e86874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:28:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a41bdb1784444f359fc2ac03d26b4064\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e3aa4a4f35d4b6da506d3c7770412c1 : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c4fc776a2b420180f1fe4a91721651 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41ec067f6acd451c8112f5857e3dd9d6 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a88988b343d46b884f44db29dbc8e65 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/367b50cc430447dabee877844679c00c : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5050f61037546dc983ee4f17bf8af7d : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321d510c51ad4c859230f37608bf4bde : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/235d769a84e0453eb9b690f2dc4db3b8 : SUCCESS in 3m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49de79ff8a58400d8458117b854a7ff5 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"f6c0ca2b66ced78e7eed5b15dd48b1fece7eebe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:29:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"353a94bc35973839b23e3678e059457be5ae006a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:37:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d71d302269e4f93a18296bbd079e8da\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ecfe58108525475e8db63cf4500f0c15 : SUCCESS in 1m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e5f28a226142739ab2c31be569b4a3 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4373743cc16541d28fb7496af9d913ee : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0f8b82d12ca443f9548575e70dedc3f : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f6c786ff894ce39cb62212f3323d41 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96fc2125fe5747feb8d443559b05155e : SUCCESS in 3m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3125bb2f398a47efbc1ae2e2d87ade44 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"bcd9b22c444f4a3f06dcd93d9371540fd7007f16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:37:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"47a0087fedf4098fbea317baf8b962d7f87f1727","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:38:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f167a1a1fe7c473b9d321a99e0f164a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2beafbb1e039400b9fafb717a6425dac : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e2f14e708ea422f83098e6ec4d6b741 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"176f48ed8671656d6d5a20e37215b849ac0978c1","revisions":{"e6ab98f14573f2e9ffea6ac5f26f20f62fc8b5eb":{"kind":"REWORK","_number":1,"created":"2023-09-18 16:51:16.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/03/895703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/03/895703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/1"}}},"commit":{"parents":[{"commit":"71b84747bf8311b4bea1c1150ef5dff844e3a744","subject":"Merge \"Get ports by \u0027binding:host_id\u0027 query filter\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/71b84747bf8311b4bea1c1150ef5dff844e3a744"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-18 16:51:00.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-18 16:51:15.000000000","tz":-420},"subject":"DNM: Test constrainting ansible version","message":"DNM: Test constrainting ansible version\n\nChange-Id: I67a3397f53e03270ac3ce540042f188991cd2bac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e6ab98f14573f2e9ffea6ac5f26f20f62fc8b5eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e6ab98f14573f2e9ffea6ac5f26f20f62fc8b5eb"}]},"branch":"refs/heads/stable/zed"},"24d7c7541f252859b6bb60fec0a0a6a0a5d8e491":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-18 19:18:04.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/03/895703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/03/895703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/2"}}},"commit":{"parents":[{"commit":"71b84747bf8311b4bea1c1150ef5dff844e3a744","subject":"Merge \"Get ports by \u0027binding:host_id\u0027 query filter\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/71b84747bf8311b4bea1c1150ef5dff844e3a744"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-18 16:51:00.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-18 19:14:26.000000000","tz":-420},"subject":"Constrain the upper Ansible version","message":"Constrain the upper Ansible version\n\nOn CI nodes, we get the latest ansible version, generally,\nsuch as 2.9.15.\n\nHowever, this code was originally developed on 2.8.x, *and*\nutilizes an older openstacksdk version which comes from our\nrequirements and stable branch upper constraints, for example\nopenstacksdk 0.101.0 being the current version on the Zed branch.\n\nHowever, Ansible 2.9 is incompatible with versions of the SDK\ngerater than 0.99.0. To make matters worse, Ansible expects\nSDK version greater than 1.0 because there are breaking changes\nfor the SDK which they navigated at the 1.0 route.\n\nSo since we\u0027re on an older stable branch, lets constrain our\ntest run ansible version so we don\u0027t end up failing the CI\njob because our base ansible is too new for the software we\nare trying to test.\n\nChange-Id: I67a3397f53e03270ac3ce540042f188991cd2bac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/24d7c7541f252859b6bb60fec0a0a6a0a5d8e491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/24d7c7541f252859b6bb60fec0a0a6a0a5d8e491"}]},"branch":"refs/heads/stable/zed"},"176f48ed8671656d6d5a20e37215b849ac0978c1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-18 19:18:47.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/03/895703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/03/895703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/03/895703/3"}}},"commit":{"parents":[{"commit":"71b84747bf8311b4bea1c1150ef5dff844e3a744","subject":"Merge \"Get ports by \u0027binding:host_id\u0027 query filter\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/71b84747bf8311b4bea1c1150ef5dff844e3a744"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-18 16:51:00.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-18 19:18:19.000000000","tz":-420},"subject":"stable-only: Constrain the upper Ansible version","message":"stable-only: Constrain the upper Ansible version\n\nOn CI nodes, we get the latest ansible version, generally,\nsuch as 2.9.15.\n\nHowever, this code was originally developed on 2.8.x, *and*\nutilizes an older openstacksdk version which comes from our\nrequirements and stable branch upper constraints, for example\nopenstacksdk 0.101.0 being the current version on the Zed branch.\n\nHowever, Ansible 2.9 is incompatible with versions of the SDK\ngerater than 0.99.0. To make matters worse, Ansible expects\nSDK version greater than 1.0 because there are breaking changes\nfor the SDK which they navigated at the 1.0 route.\n\nSo since we\u0027re on an older stable branch, lets constrain our\ntest run ansible version so we don\u0027t end up failing the CI\njob because our base ansible is too new for the software we\nare trying to test.\n\nPresently planning to backport this on stable branches so we\ncan unbrick IPA stable branches.\n\nChange-Id: I67a3397f53e03270ac3ce540042f188991cd2bac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
